)]}'
{"id":"openstack%2Fnova-specs~662881","triplet_id":"openstack%2Fnova-specs~master~I9aadc5e653e9f5af0d47028ffe74eaf439658950","project":"openstack/nova-specs","branch":"master","topic":"bp/openstacksdk-in-nova","hashtags":[],"change_id":"I9aadc5e653e9f5af0d47028ffe74eaf439658950","subject":"Spec: Use OpenStack SDK in Nova","status":"MERGED","created":"2019-06-03 23:41:47.000000000","updated":"2019-07-03 00:30:21.000000000","submitted":"2019-07-02 22:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":51,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662881-1562107670751-cbb5daaf","meta_rev_id":"e5b0b51a6f21a6d742cca1e492207b5aed079d6a","_number":662881,"virtual_id_number":662881,"owner":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-01 18:48:07.000000000","_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-07-02 22:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-07-02 14:55:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-06-24 19:44:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-07-02 16:43:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-28 16:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-01 07:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-27 16:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-07-02 22:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-06-28 21:54:28.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 19:08:56.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-06-04 20:03:31.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-24 12:30:05.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-24 19:44:25.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-27 16:45:20.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-06-28 21:54:28.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-07-01 07:53:43.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-07-02 14:55:47.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-07-02 16:43:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-02 22:27:47.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-07-02 22:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d696ce6e6262c8cb8ba08c420fb4546181fe4005","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-03 23:41:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"224feb34743a8e2deb7da9a2705785061df6c14f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 23:52:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/1/check/openstack-tox-docs/cd8df65/html/ : SUCCESS in 8m 25s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/1/check/openstack-tox-pep8/fad2f3a/ : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f9b42447bdc6d6877b5a2958e9368928fface8a6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-04 11:00:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI request to look into the functional testing impact of the proposal.","accounts_in_message":[],"_revision_number":1},{"id":"caaf7f086c1fe0fabba058178154c3da45976fc9","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-04 11:27:11.000000000","message":"Patch Set 1:\n\nI\u0027d love to see this happen, but it seems like there are a fair number of dependencies to satisfy before this can have a significant impact throughout nova:\n\n* switching the SchedulerReportClient from ksa to sdk is (as described) straightforward (because underneath it is just http requests) and a good incremental step\n* for cinder, neutron, glance it seems like a lot of work must happen in either the SDK or Nova or both to create any missing code that is doing \"just the http requests\". How much is missing?\n* since we presumably don\u0027t want to duplicate work in nova and the SDK, doing that coding in the SDK seems like the right approach\n\nWhich suggests that there can be three phases to this work:\n\n* the placement adjustments in nova\n* the SDK work\n* incrementally incorporating that SDK work\n\nIf some of the SDK work is already done, then steps two and three can be interleaved.\n\nThat\u0027s pretty much what this spec says, but in order for the spec to be actionable in any immediate sense, it might be good to split it in two: what can be done now, what will be done later.","accounts_in_message":[],"_revision_number":1},{"id":"dcf58d2b5e6a30e9e1781e2e8af457fe5d8b767d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-04 19:08:24.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI\u0027m going to feed off of Chris\u0027s comments a bit and ask that the staging and dependencies be explained in a bit more detail.\n\n(A) Assuming that $service is already set up to understand ksa adapter opts (which I think is true for everything but cinder - wip at [1]) we can introduce the sdk \"plumbing\" (just building a Connection and Proxy) for each $service with no cost other than a dependency on [2]. At its most basic, this step can use the Proxy as a drop-in replacement for whatever the Adapter is doing today - which is usually discovering an endpoint to pass to the ${service}client constructor.\n\n(B) From there we can incrementally, one operation per commit, start replacing calls through ${service}client with calls through the SDK - *for operations that are already supported by the SDK* - like you\u0027ve been doing for ironicclient [3].\n\n(C) At some point you may run out of things that are already supported by the SDK, whereupon the nova side work for a given service would stall while those operations are added to the SDK side. Add operation in the SDK, add the cutover in nova, repeat.\n\n(D) Once all the operations are going through the SDK, remove the ${client} plumbing and dependencies.\n\nSDK and ${service}client calls coexist during (B) and (C) - and that\u0027s fine and dandy, even if that\u0027s the state of affairs at a release boundary.\n\nWhich is another thing that could be mentioned: this doesn\u0027t need to all land in one release, for any one or subset or all services. Because deployer/operator/admin/upgrade impact is nil at all stages.\n\nUntil we start talking about clouds.yaml, which should be at least mentioned in this spec as a direction we want to head, though out of scope here.\n\n[1] https://review.opendev.org/#/c/655985/\n[2] https://review.opendev.org/#/c/643601/\n[3] https://review.opendev.org/#/q/topic:bp/openstacksdk-in-nova+(status:open+OR+status:merged)+project:openstack/nova","accounts_in_message":[],"_revision_number":1},{"id":"a6848f85ece5aa6c9d098e26496381f601498e1d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 19:33:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbaa69eb8be13422f9b5a156835feb7f28dc7af2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 19:34:28.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"db541e7dd49326cae9e989e125644b7590346dbe","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 19:34:32.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"96f077358df987e2f44e8ff36f86948e4f0d31a9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-04 19:58:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb22f13404ce453d9a990fca873d5d4052962db4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-04 20:03:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c5d9aee78dc28bdc35f047acb3f280516bf8d4b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-04 20:12:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f19affd469e1e55831203a41125ec80e3f64fadb","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-04 21:25:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cb8ba4a1f2ac1f2ed78e18c0da2180e8c3cc5be8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 21:39:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/2/check/openstack-tox-docs/d1d5a0f/html/ : SUCCESS in 7m 56s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/2/check/openstack-tox-pep8/b105ede/ : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":2},{"id":"af6f4ef05ee6199742d16916d5ea254f0611c859","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-04 23:16:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ec9bbae465ed8e0fd8409a7feec8fb2ca782fa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 23:25:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/3/check/openstack-tox-docs/2a735cf/html/ : SUCCESS in 7m 32s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/3/check/openstack-tox-pep8/0e2aa6f/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":3},{"id":"266c73338d88b657a3afb8578255cdf7c4d9990f","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-05 03:02:17.000000000","message":"Patch Set 3: Workflow-1\n\nThanks for the feedback. I incorporated some of it and will be working on the rest in this spec and moving some out to another spec.","accounts_in_message":[],"_revision_number":3},{"id":"86d1b3a57115d46127e67bb3dbc3b4db32fc9d44","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-21 00:42:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"acf63f667f112ad2b9b0cf28a21979d2ae43d609","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-21 00:43:17.000000000","message":"Patch Set 4: Workflow-1\n\nThis is still a WIP and needs another update.","accounts_in_message":[],"_revision_number":4},{"id":"741f8f04ff902a73389e97eddd25a17154bdfd42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 01:13:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/4/check/openstack-tox-docs/5e3917f/html/ : SUCCESS in 7m 23s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/4/check/openstack-tox-pep8/0a1e5f3/ : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"0f4d5b9ec57adb73524a89bdd455a212fe3e5b38","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-21 23:47:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0cc9162a891a84f69533beab0b675b557fc533f4","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-21 23:48:37.000000000","message":"Patch Set 5:\n\nI still need to work on documenting the testing impacts, but will need some help understanding them first. In the meantime, the rest of the spec is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"13c263f0b68b7b191ecafb73eff46cbb4bca633e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 23:58:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/5/check/openstack-tox-docs/cbab078/html/ : SUCCESS in 8m 14s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/5/check/openstack-tox-pep8/2438921/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":5},{"id":"c19779ae0fe2619c4e9a91b22ffda7f3cd85c7d9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 12:50:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(16 comments)\n\nClean and simple, thanks Dustin. I think this is very close.\n\n-1 to ack that the testing impact section still needs to be filled in (provided some seed text for that). Other minor/nonblocking issues noted inline.","accounts_in_message":[],"_revision_number":5},{"id":"5ca65d74fde541a3aeecbbaa1a2efe40c8a343d7","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-24 16:26:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0522513f69ac44ade213769f5370af0a0d7aff07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 16:35:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/6/check/openstack-tox-docs/4dc58c2/html/ : SUCCESS in 7m 47s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/6/check/openstack-tox-pep8/0937ef3/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"a87dd7e2635e8585432b83ba0e89c585bc411d2f","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-24 17:19:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4dfc66e164b697ebf1b0fcae74af1cdc6583d28a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 17:34:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/7/check/openstack-tox-docs/e7c4616/html/ : SUCCESS in 7m 34s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/7/check/openstack-tox-pep8/873ba1c/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":7},{"id":"597d060f3a9fc8f0f93aea52f78ce90bea0161b3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 18:06:38.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)\n\nThis looks ready to me, other than a couple of tiny paperwork issues.\n\nI\u0027ll recuse myself from approving since I\u0027ve been materially involved in the concept and design.","accounts_in_message":[],"_revision_number":7},{"id":"9f1fa9c9af079957f06e1919b432e5905768eb6e","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-06-24 19:30:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f20da9b7a5961f8cefd836211d2540a1a9243146","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 19:41:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/8/check/openstack-tox-docs/a5fbbe7/html/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/8/check/openstack-tox-pep8/63c79ea/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":8},{"id":"c22177bfbd87aa039db3754352052bf0ddad6e24","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 19:44:25.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"2790429746d0043a9555f903a9f8bd0c2bd05ab0","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-27 16:45:20.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)\n\nNice, I was +2 this at the PTG, and this spec contains what I expected to see.","accounts_in_message":[],"_revision_number":8},{"id":"f15b0f32c89a14840e9af1d8d0698b826d2d1c10","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 16:55:23.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nThanks John.","accounts_in_message":[],"_revision_number":8},{"id":"4b2b1753741aead7e6bdd468d51cca8bc3ddf8cd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-28 01:09:57.000000000","message":"Patch Set 8:\n\nRemind / ping / harass me to come back to review this.","accounts_in_message":[],"_revision_number":8},{"id":"974e20cbd7535166168fa8025ed41c87bf0ea347","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-28 16:10:54.000000000","message":"Patch Set 8: Code-Review+2\n\n(14 comments)\n\nI\u0027m OK with this in general and have left lots of comments about additional motivations for this with examples if you wanted to follow up, plus a few nits and some personal whining.\n\nI\u0027m only going to +2 for now because this is a nova-wide change and as such I think should at least have time (maybe a week? or until at least the EOD on July 2 for the 3rd and final spec review sprint for Train?) for other cores to make sure they are aware and are on board with this.","accounts_in_message":[],"_revision_number":8},{"id":"8c9f745a62ebf58549bfa2f4688ba1d374ede629","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-28 16:17:08.000000000","message":"Patch Set 8:\n\nPosted to the ML:\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-June/007436.html","accounts_in_message":[],"_revision_number":8},{"id":"b685653cacae98d874f96edc22aab336c4c36443","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-01 07:53:43.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"e2066b50cfd5c6ea4c2a695490abcd6b11a8bd55","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-07-01 18:48:07.000000000","message":"Patch Set 8:\n\nThanks for reviews.","accounts_in_message":[],"_revision_number":8},{"id":"7b36489f6dd058fc18f78210652b036433c20307","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-07-02 14:55:47.000000000","message":"Patch Set 8: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"53cdc295e64e72c0c83f767e5fb89cc8661ca8ae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-02 15:05:27.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f9a65fabb806c99180a0c5babddc17df35c3445c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-02 16:43:44.000000000","message":"Patch Set 8: Code-Review+1\n\nNot an in-depth review, but this looks sane and is something that should be done","accounts_in_message":[],"_revision_number":8},{"id":"e0f99dc94d98c1f32f1e685cc50b6d9c040ff89f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-02 22:27:47.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)\n\nI would like to see a follow up with some of the comments addressed, like the other dog-fooding use case I mentioned and which Sean highlighted, but otherwise the advertised time has passed for us to approve this (no nay-sayers so assuming others have abstained).","accounts_in_message":[],"_revision_number":8},{"id":"a11fa804739e11cf10263290e5623adbd54d2e51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 22:27:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7ff5abee123b58fe0600d55f5e96f1a3649eed82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 22:47:50.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/662881/8/gate/openstack-tox-docs/b9c5751/html/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/81/662881/8/gate/openstack-tox-pep8/fef12f9/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":8},{"id":"c990834e6386e122e44a1e0b8b2d4a0a18bd0901","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 22:47:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"dc1d57be12f58d29f489f1af37c456c5229975e6","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-07-03 00:30:21.000000000","message":"Patch Set 8:\n\n\u003e I would like to see a follow up with some of the comments\n \u003e addressed\n\nI plan on doing this relatively soon. Thanks for all the great feedback and help everyone!","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"13184aea4b7f54c4c980334441e5ac342dac9662","revisions":{"cd7d18a3c4cce1e0e1cc052e3cc1ab6ef71eee51":{"kind":"REWORK","_number":1,"created":"2019-06-03 23:41:47.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/1"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:41:46.000000000","tz":-420},"subject":"Spec: Use OpenStack SDK in Nova","message":"Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cd7d18a3c4cce1e0e1cc052e3cc1ab6ef71eee51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cd7d18a3c4cce1e0e1cc052e3cc1ab6ef71eee51"}]},"branch":"refs/heads/master"},"962b713ba913d2966d9d0c0a701756050509f782":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-04 21:25:52.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/2"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-04 21:25:19.000000000","tz":-420},"subject":"WIP: Spec: Use OpenStack SDK in Nova","message":"WIP: Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/962b713ba913d2966d9d0c0a701756050509f782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/962b713ba913d2966d9d0c0a701756050509f782"}]},"branch":"refs/heads/master"},"095f22597c7638be48612dcf22002da4e31d127f":{"kind":"REWORK","_number":3,"created":"2019-06-04 23:16:14.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/3"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-04 23:15:55.000000000","tz":-420},"subject":"WIP: Spec: Use OpenStack SDK in Nova","message":"WIP: Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/095f22597c7638be48612dcf22002da4e31d127f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/095f22597c7638be48612dcf22002da4e31d127f"}]},"branch":"refs/heads/master"},"9aa60f7353427fbe3aa5981ca49941596969bcaa":{"kind":"REWORK","_number":4,"created":"2019-06-21 00:42:51.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/4"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-21 00:42:19.000000000","tz":-420},"subject":"WIP: Spec: Use OpenStack SDK in Nova","message":"WIP: Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9aa60f7353427fbe3aa5981ca49941596969bcaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9aa60f7353427fbe3aa5981ca49941596969bcaa"}]},"branch":"refs/heads/master"},"f3822b7bbc672b5de7fa358f1ef22509e5406435":{"kind":"REWORK","_number":5,"created":"2019-06-21 23:47:43.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/5"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-21 23:47:25.000000000","tz":-420},"subject":"Spec: Use OpenStack SDK in Nova","message":"Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f3822b7bbc672b5de7fa358f1ef22509e5406435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f3822b7bbc672b5de7fa358f1ef22509e5406435"}]},"branch":"refs/heads/master"},"fda3fba4c504fa8b3cc38deb28c026860c2a1263":{"kind":"REWORK","_number":6,"created":"2019-06-24 16:26:14.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/6"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-24 16:26:12.000000000","tz":-420},"subject":"Spec: Use OpenStack SDK in Nova","message":"Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fda3fba4c504fa8b3cc38deb28c026860c2a1263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fda3fba4c504fa8b3cc38deb28c026860c2a1263"}]},"branch":"refs/heads/master"},"b18a8bf66ed87bf8982a41d9801aa15cda151f9a":{"kind":"REWORK","_number":7,"created":"2019-06-24 17:19:22.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/7"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-24 17:19:02.000000000","tz":-420},"subject":"Spec: Use OpenStack SDK in Nova","message":"Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor ${service} in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b18a8bf66ed87bf8982a41d9801aa15cda151f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b18a8bf66ed87bf8982a41d9801aa15cda151f9a"}]},"branch":"refs/heads/master"},"13184aea4b7f54c4c980334441e5ac342dac9662":{"kind":"REWORK","_number":8,"created":"2019-06-24 19:30:30.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/81/662881/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/662881/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/662881/8"}}},"commit":{"parents":[{"commit":"090d47bf808b451407b84744ad5cc6a06b5cb693","subject":"Merge \"Fix a typo of \u0027keystoneauth1\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/090d47bf808b451407b84744ad5cc6a06b5cb693"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-03 23:39:54.000000000","tz":-420},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-06-24 19:26:53.000000000","tz":-420},"subject":"Spec: Use OpenStack SDK in Nova","message":"Spec: Use OpenStack SDK in Nova\n\nThis is a proposal to get rid of Nova\u0027s use of python-${service}client\nfor $service in glance, ironic, neutron, cinder, etc and replace them\nwith the OpenStack SDK.\n\nChange-Id: I9aadc5e653e9f5af0d47028ffe74eaf439658950\nImplements: blueprint openstacksdk-in-nova\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/13184aea4b7f54c4c980334441e5ac342dac9662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/13184aea4b7f54c4c980334441e5ac342dac9662"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
