)]}'
{"id":"openstack%2Fnova-specs~663563","triplet_id":"openstack%2Fnova-specs~master~Ied367a32d3785b29c4ba7b64a04c855c407a38e7","project":"openstack/nova-specs","branch":"master","topic":"bp/composable-flavor-properties","hashtags":[],"change_id":"Ied367a32d3785b29c4ba7b64a04c855c407a38e7","subject":"Add composable flavor properties","status":"ABANDONED","created":"2019-06-06 09:35:43.000000000","updated":"2020-02-05 06:56:15.000000000","total_comment_count":144,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"160a554ed9e7f661f66b3a1af907b0249a367f5a","_number":663563,"virtual_id_number":663563,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-20 09:20:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"date":"2020-02-05 06:54:34.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":-2,"date":"2019-12-20 09:04:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-2,"date":"2019-12-20 09:04:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 04:33:59.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-07-25 05:59:41.000000000","updated_by":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"reviewer":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"state":"REVIEWER"},{"updated":"2019-07-25 06:01:33.000000000","updated_by":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"reviewer":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"state":"REVIEWER"},{"updated":"2019-07-25 06:01:46.000000000","updated_by":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"reviewer":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"state":"REVIEWER"},{"updated":"2019-07-25 06:01:46.000000000","updated_by":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"reviewer":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"state":"REVIEWER"},{"updated":"2019-07-25 06:04:08.000000000","updated_by":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"reviewer":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"state":"REVIEWER"},{"updated":"2019-07-25 06:20:38.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2019-07-25 09:11:49.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"state":"REVIEWER"},{"updated":"2019-08-06 02:05:18.000000000","updated_by":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"reviewer":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"state":"REVIEWER"},{"updated":"2019-11-04 17:04:24.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-11-04 17:07:23.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-14 08:09:09.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-18 23:35:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-11-20 10:10:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-12-20 09:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff0d3091f216863d85340e34f3a21c0942d7b55f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-06 09:35:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec5a13b13e3128d207e69168b922e559b107bc64","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-06 09:37:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eb5d15b6e68d84127b639acf43b4e535e19c8be0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:52:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/663563/2/check/openstack-tox-docs/0ebc9da/html/ : SUCCESS in 7m 27s\n- openstack-tox-pep8 http://logs.openstack.org/63/663563/2/check/openstack-tox-pep8/79edb88/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"9164eff429ae1c098b73e2d521ea5d81f3cf54ea","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 22:26:05.000000000","message":"Patch Set 2: Code-Review-1\n\nAs I started reading this, I *thought* you were going to say:\n\nCreate a separate place to store chunks of flavor metadata, so you could effectively \"modularize\" extra_specs. The \"modules\" would be groupings of key/value pairs that are logically associated with each other, but not necessarily associated with other \"modules\". Then either by listing them by name/ID in the flavor, or specifying them directly to the boot request, you could customize your instance without having to repeat the same information in multiple flavors, or have \"flavor explosion\" where many almost-identical flavors exist that only differ in small ways.\n\nWhich I thought would be a pretty nice idea.\n\nBut all I see here is a table where arbitrary key/value pairs can be stored. I don\u0027t see any description of how those key/value pairs might be grouped together, or how you associate them with a flavor and/or instance.\n\nI think there\u0027s a seed of potential here, but it definitely needs further development.","accounts_in_message":[],"_revision_number":2},{"id":"cf571bac0927bdc56bf7a19e68f70f6ef9e1ac48","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-26 13:09:56.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Create a separate place to store chunks of flavor metadata, so you \n \u003e could effectively \"modularize\" extra_specs. The \"modules\" would be \n \u003e groupings of key/value pairs that are logically associated with each \n \u003e other, but not necessarily associated with other \"modules\". \n \n This is the central idea of spec :P \n\n \u003e But all I see here is a table where arbitrary key/value pairs can\n \u003e be stored. I don\u0027t see any description of how those key/value pairs\n \u003e might be grouped together, or how you associate them with a flavor\n \u003e and/or instance.\n\nYeah, those key/value pairs need to create an associated with the flavor, IMO, ``flavorid`` is ok. \n\n \u003e I think there\u0027s a seed of potential here, but it definitely needs\n \u003e further development.\n\nThanks Eric Fried.\nWhether it\u0027s from our customers or our own OpenStack environment, we think that \"flavor\" is too strongly related to \"extra_specs\".\nSometimes we need to create a lot of flavors in the environment, but their memory, disk and the CPU are the same, this is a very annoying thing.\nTherefore, long-term development, it is necessary for the flavor to make changes. I hope that you and others can give more comments on **this** spec, to form a perfect flavor modification plan.","accounts_in_message":[],"_revision_number":2},{"id":"9015d21d40eb8c667fe346a091ee40f75607b64a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-26 16:58:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e4bee6a9188efb8786984c5f5a06424c207462f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-27 00:54:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00ff35317472c8ca2a5a723135878ed2a4f2723f","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-27 04:33:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6311ad28b68a854f71997a3d813935e765aabd13","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-27 07:55:39.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nThanks Alex.\nThat will be addressed.","accounts_in_message":[],"_revision_number":2},{"id":"ed2e9e32f8b9ee8d4b8a7991e2aece7c76ad8506","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-27 10:17:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73dba31b4b885a5bcba6f9f8addac9392ccc5d21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:29:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/663563/3/check/openstack-tox-docs/9532879/html/ : SUCCESS in 7m 38s\n- openstack-tox-pep8 http://logs.openstack.org/63/663563/3/check/openstack-tox-pep8/2f514db/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8acfac1b0fac6075c3230916b42a95ce8c4061b5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 19:34:17.000000000","message":"Patch Set 3: Code-Review-1\n\nI still don\u0027t see the APIs for groups (modules). (IMO we *only* need APIs for groups, not for individual keys. By far I would expect the most common use case to be groups; if you really need a single key, you can always make a group-of-one.)\n\nI\u0027m also still not seeing discussion of how to associate the groups with flavors.\n\nBut mainly:\n\nI don\u0027t think this is realistic for Train. Could you please propose it to the backlog/ subdirectory [1] instead?\n\n[1] http://specs.openstack.org/openstack/nova-specs/specs/backlog/index.html","accounts_in_message":[],"_revision_number":3},{"id":"4f1fc903916ce8dc32cb8e92a625bbd59865e138","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-28 10:36:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"31fe1afe6ebc2b085f89ac1a5c5c82838e240c77","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-28 10:37:04.000000000","message":"Patch Set 4:\n\nThanks Eric Fried.\nI discussed it with my OpenStack development team. As specified by the spec, ``flavor_group`` will be associated with the instance when the instance is created, no longer associated with the flavor, and completely decoupled.\n\nMove to this spec to the backlog.\nFor the current design, please make suggestions to be more perfect, thank you.","accounts_in_message":[],"_revision_number":4},{"id":"9e7d2b395d745b0ff61e4dd6684dd95ad74ad872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:47:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/663563/4/check/openstack-tox-docs/afc3c90/html/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/63/663563/4/check/openstack-tox-pep8/5d9f711/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"5c6a9227b1d1b5b52ea63506597afe6b64e8c7d6","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-07-25 05:59:41.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"6fafc223d9d9ed502d6117f47e404538271d9157","author":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"date":"2019-07-25 06:01:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"07b054c8d15f94f42836c87c30f2f0f6ef956fea","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-07-25 06:01:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6e7da378373287dab288c4e7f16285f284e3389d","author":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"date":"2019-07-25 06:01:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ca5b485e466a99f4962ad10a0f9fe64d680e4daf","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-07-25 06:04:08.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b84e40cac909b9c87c27373400ec874f5d07549e","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-07-25 06:20:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"323f9538c68bb694a54c3296a67ccbd4b1acb971","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-07-25 09:11:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ecf63278f499474fa812e13f525b72cd8dac0ac7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-25 13:35:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"264601616463a4c2bb3e20892432e643a78a8264","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-30 03:25:33.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"67e169e30562cba13040b8eb8ed7ff7ae6c68a09","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-30 03:25:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5262c7f2021c9eec1ece7230e8033551125a4a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 03:35:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/663563/5/check/openstack-tox-docs/255d919/html/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 http://logs.openstack.org/63/663563/5/check/openstack-tox-pep8/10b3472/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"2b4c66ccc1e2ab649003a25c89ae7b92c2bb744c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-28 11:16:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"2160d0fc52f8c41f0da09b47447755ea7ebdbf62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:29:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1986f8c3e34ddfb240de1ddde1e2a4 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716490e42ade40959804433b88758101 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":6},{"id":"7a3b5977a09b24b333fcc010ff943ec767116db7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-31 19:15:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)\n\nI like this concept, but I\u0027m not sure it\u0027s going to be worth the effort-to-value ratio.\n\nSome nontrivial concerns within.","accounts_in_message":[],"_revision_number":6},{"id":"61e6d83ca6283cd6b39e769c1aa643cc27291225","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-10-31 20:06:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(13 comments)\n\nI think this is lacking a ton of detail, or I\u0027m missing it. I\u0027m very fuzzy on what this looks like when it\u0027s done, and definitely how we get from here to there. I\u0027m also skeptical of the benefit-vs-complexity of what is described here, as well as whether or not it aims to be too flexible to satisfy the more cloudy operators that need very strict control over what instances look like.\n\nMaintaining a bunch of flavors to provide multiple combinations of, say, memory amounts with numa topologies, is somework, but it\u0027s very (very) scriptable and composition of those things is pretty easy to do in operator tooling that defines them.\n\nSo, I don\u0027t think I like where this is headed, but I also think it\u0027s missing too much detail for me to even know what that destination is exactly. Hopefully my questions inline will help flesh some of that out.","accounts_in_message":[],"_revision_number":6},{"id":"302815b79c80c12487e1f2b7856570bcd76a0b3f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-03 15:29:51.000000000","message":"Patch Set 6:\n\n(13 comments)\n\nThanks Dan, Eric.\n\nI responded to the question and hope that you can learn more details.","accounts_in_message":[],"_revision_number":6},{"id":"11eba38225fe96848bd4b55f3970062fe00c1730","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-03 15:30:15.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"c96e18fb4f9b5cf54696b3ca1b2faee920bf6858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 15:36:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322297e2867b47ef98a8bf420172eb53 : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ddbd1b511d94d548cbb36777f46ae7a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":7},{"id":"4d029f2520e505c0b67d336007d5a3a7591690bf","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-04 09:06:27.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2d028c397ef730cc89f2f4f51c14328a8882b1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 09:15:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938d11e84bd84e1ba230b975e044618d : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993b552ad1d749f5bd7ef1e7a7a0de0d : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":8},{"id":"8ea79a6734b7f69e73f53517d2047779e230127b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-04 15:01:46.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"92b7867756c17787e1a312cec37b86e37605da3f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-04 15:06:18.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nI\u0027m tempted to -2 this, as that\u0027s where I\u0027m sitting right now but will hold to a -1 for the moment to avoid looking like I\u0027m completely shutting it down. However, I definitely feel that we should NOT do this.","accounts_in_message":[],"_revision_number":8},{"id":"77ec1ce0ec7b0648782670d1f0c98587294f3e4b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-04 16:45:53.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)\n\nThere are still major chunks of detail missing here. But it may not be worth going to the trouble of filling them in, because it is looking like there won\u0027t be a lot of support for this idea.","accounts_in_message":[],"_revision_number":8},{"id":"096fe8c46a119f7c988ec04b348d051941f42645","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-04 17:04:24.000000000","message":"Patch Set 8: Code-Review-2\n\n(1 comment)\n\n\u003e There are still major chunks of detail missing here. But it may not\n \u003e be worth going to the trouble of filling them in, because it is\n \u003e looking like there won\u0027t be a lot of support for this idea.\n\nYeah, I guess I\u0027m going to go ahead and -2 this to make it clear that I\u0027m not okay with this without major and substantial changes. It will take a lot to get from where this is today to a point where I think it\u0027s worth the effort, complexity, and change to the operator and user experiences.","accounts_in_message":[],"_revision_number":8},{"id":"d13792242ca058a19a98ae65849d365cf28dae23","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-04 17:07:23.000000000","message":"Patch Set 8: Code-Review-2\n\nI agree: even with backward-compatible support for flavor extra specs, there\u0027s too much complexity to make this worthwhile.\n\nThanks for the idea, please keep them coming!","accounts_in_message":[],"_revision_number":8},{"id":"061f202b83e0178ffc1d7307e7d9a7856aa7b446","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-06 06:50:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0d8385b3f85516679d3b1bdb636fdf0db47686df","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-10 15:47:45.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c3908ccc5e29a846388819f66770b77aac024e72","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-10 15:53:07.000000000","message":"Patch Set 9:\n\nAt the Shanghai PTG meeting, we reached the following agreement on this SPEC:\n* Do not try to change the model of the flavor in nova code and in the Db.\n* Define a \"ComposableFlavorBit\" (bikeshed on the name please) REST API entity that can hold any kind of flavor bits (extra specs, normal flavor fields), propose some format in the spec for it. This entity can only be created by the Admin by default.\n* Extend the server create REST API to allow the end user to specify what \"ComposableFlavorBit\"s she wants to add to the \"base\" flavor she used in the create request.\n* The nova api then merges the \"ComposableFalvorBit\"s with the base flavor and embed the resulting flavor object into the instance.\n* Do a similar thing for resize\n\nUpdate on this SPEC, from PTG meeting discussion results [1] and ML [2].\n[1] https://etherpad.openstack.org/p/nova-shanghai-ptg\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-November/010643.html","accounts_in_message":[],"_revision_number":9},{"id":"391f279dec1bc977d7ff040b35507c5c47491464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 15:56:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d9f8bd365b47c3ab86299569a2ab64 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a229709626d042cb86cc0c93965f2dd5 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":9},{"id":"973421c959cce1aa88189510c65abdc5d6cc3e3f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-10 15:59:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d6edcbba8873a6572e772e4bf163789c0f12d6c7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-11-14 08:29:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(16 comments)\n\nThis spec needs more work.","accounts_in_message":[],"_revision_number":9},{"id":"f386c7bbc333bfdb9851a81f72c3ca85c5b46ee8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-14 08:55:58.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c5d00c43a9b85192c2f1ba14cd19c8dcf6b49bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 09:04:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0650f98ed589497393ed90682dd22431 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87936faa8ce1407784bf5bc0ee0f5602 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":10},{"id":"b39b4840aea83ce1e212d28317fa2cfe59206d2c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-14 11:02:00.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5bf67e6b2380552ab22decc6cfb58b89d1b7d18f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-11-14 12:30:25.000000000","message":"Patch Set 10: Code-Review-1\n\n(8 comments)\n\ni think this has a long way to go.\ni innitally though you were going a different direction untill i got to the server create line.\n\ni have mofied some of my comments retroactivly to reflect that but i would like to see more detail in this for the usecase you want to solve as im not sure the solution you are proposing is the right one for that problem.","accounts_in_message":[],"_revision_number":10},{"id":"2f8d09c96c3866a301b513e4afc1c9875b733914","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-15 01:52:06.000000000","message":"Patch Set 10:\n\n(14 comments)\n\nThanks gibi.","accounts_in_message":[],"_revision_number":10},{"id":"8b910280e7550755a753c210839431bc71fc6ae8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-15 03:47:11.000000000","message":"Patch Set 10:\n\n(7 comments)\n\nThanks sean.\nWith the db table design I have a different opinion from you, please review the latest patch, if you need improvement, please leave comments.","accounts_in_message":[],"_revision_number":10},{"id":"6da9c1762b7fcc42e8be49121775e72a39179941","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-15 05:41:05.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3b2468453419be5dd4ba1ea8ef7092a677842977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 05:51:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b47c09042a94d4b9461d777420d4462 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf9c5e73f23470ea386c523bbfd70eb : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":11},{"id":"13493e1279e510f1caaeccfb7f902c09b4c813c2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-15 06:02:04.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"d270d1f5cb87982d114991b83c5854bdf606678c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-15 06:45:33.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7cffbe67e773d2d9ef33992db90c22f3555ea7d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 06:56:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/811f37657ce2419da5d36e6a4f3f3714 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a83122d71e384844b43455e281e45395 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":12},{"id":"a20bf7c2ccf1624686537ea37ddfa72c42c798f4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-11-18 22:21:53.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"cc3c6898fdc7ee4c9be093c05a43fc030b582917","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-11-18 23:35:18.000000000","message":"Patch Set 12: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e8794735480d8d6194b963cc5a0ff0b7cae434d5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-11-19 11:19:53.000000000","message":"Patch Set 12:\n\n(8 comments)\n\nI\u0027ve read back to Dan\u0027s and Eric\u0027s comments to see if the current -2s are still valid or not. What problems I see from those comments:\n\nFundamental problems:\n* Upgrade path for old flavors to composable flavors. \u003d\u003e The proposal discussed on the PTG has the advantage that is does not change any legacy behavior if the admin does not create any composable bits. Also it does not change any existing db model or RPC message. The API addition can be covered in a microversion. So I think the upgrade is pretty OK.\n\n* Conflicting composable bits \u003d\u003e To resolve this I suggest that as a first step reject everything where base flavor or any composable bits in the request have the same extra spec. \n\n* Restricting composability, basically answering the question: Which base flavor can be combined to which composable bit? \u003d\u003e Why not allow every combination at the first step. This feature would be opt in for the deployers anyhow. Later the composable bit can have a list of base flavor ids to allow restricting the composability.\n\nProcedural problems\n* Who will implement this? I guess it it the spec author\n* Who ill review this? If the implementation is limited to the API level and api db level (for the composed bit model) then I don\u0027t think this will be a hair review (no rpc, no virt driver changes). So sure you can sign me up to review it.\n\n@Dan, @Eric: Do I understood your problems properly behind your -2s? Can we make steps forward with this or you still feel that this does not worth the effort?\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":12},{"id":"51d0de845ad1a7a96606887bc811758fdc799623","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-19 15:09:54.000000000","message":"Patch Set 12:\n\n\u003e @Dan, @Eric: Do I understood your problems properly behind your -2s? Can we make steps forward with this or you still feel that this does not worth the effort?\n\nThanks for outlining the design points, gibi. But my -2 was already assuming these things were the intent (esp. \"we\u0027re not going to change the way flavors work\" and \"this is 100% opt-in\"). My -2 was purely \"this isn\u0027t worth the effort\". IMO we have enough more important things to do, some of which have been waiting for years, that we shouldn\u0027t expend resources on this feature.\n\nIf there is enough of a dissenting consensus, I will \"upgrade\" my vote and let this merge with my -1. But I want to wait to do that until the design text is ironed out so I don\u0027t have to keep re-voting each PS.\n\nNote, however, that part of my -2 was to signal \"don\u0027t bother spending a bunch of effort getting the design text ironed out, because this is unlikely to fly for non-design reasons\".","accounts_in_message":[],"_revision_number":12},{"id":"18dbe50a43652fdea6b6284fd2b3900a61f71503","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-11-19 15:19:02.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7f95aee8e9b6f8e68c881d1d0a985f2a7d398b16","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-19 16:36:24.000000000","message":"Patch Set 12:\n\n\u003e @Dan, @Eric: Do I understood your problems properly behind your\n \u003e -2s? Can we make steps forward with this or you still feel that\n \u003e this does not worth the effort?\n\nI actually don\u0027t want compose-able flavors at all, so I\u0027m at least -1 on that as a thing. I understand some do, and so I\u0027m definitely -2 on any proposal that moves to a world where things are only composed (as this spec was initially). IMHO, in order for something like this to proceed, we *must* have control over which flavors can be composed with which bits. Starting off with no restriction is not a better plan to me.\n\nI think the underlying goal of this spec is to supposedly reduce the management overhead of defining flavors with various pieces of advanced extra specs, not just to provide the compose-able nature altogether. I challenge the assertion that this is a good solution to not having some scripts to automate that task and don\u0027t want to see us adopt something which I (and seemingly Eric) expect will bring significant complexity.\n\nIf a deployer has a private cloud where everything should be compose-able by users (as this spec seemed to assert initially), why not just open up the policy to let users create flavors? Then they can create what they want, delete it them if/when they\u0027re finished, and use ansible or bash to template things out?","accounts_in_message":[],"_revision_number":12},{"id":"c3aa89873b4f8d1e68fbd3fe60279950b1dab81a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-20 04:42:58.000000000","message":"Patch Set 12:\n\n(8 comments)\n\nThanks Sean. gibi and all reivewers.\n\nAgree gibi, this step needs to be simplified. In this SPEC, the concept of group/roules will be removed. The first step is to start from key/value and make it an optional feature of the operator.\n\nIn any case, I should update SPEC to the latest and live up to the reviewers who made suggestions.","accounts_in_message":[],"_revision_number":12},{"id":"ec5a4efceb962daa8643412582b73c2e1ac8aa5a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-20 04:58:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e721efe3a9ff2e85cf6e5346071ff0c9f0ff8f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 05:07:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33d6f82a1894df289c5b758c5780d50 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b966b5eddc734aa4a1fd30cd6c0c23f6 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":13},{"id":"8ee5a4241a393a2892f320bb59b645ee04a1bdb4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-11-20 09:26:03.000000000","message":"Patch Set 13:\n\n\u003e \u003e @Dan, @Eric: Do I understood your problems properly behind your\n \u003e -2s? Can we make steps forward with this or you still feel that\n \u003e this does not worth the effort?\n \u003e \n \u003e Thanks for outlining the design points, gibi. But my -2 was already\n \u003e assuming these things were the intent (esp. \"we\u0027re not going to\n \u003e change the way flavors work\" and \"this is 100% opt-in\"). My -2 was\n \u003e purely \"this isn\u0027t worth the effort\". IMO we have enough more\n \u003e important things to do, some of which have been waiting for years,\n \u003e that we shouldn\u0027t expend resources on this feature.\n \u003e \n \u003e If there is enough of a dissenting consensus, I will \"upgrade\" my\n \u003e vote and let this merge with my -1. But I want to wait to do that\n \u003e until the design text is ironed out so I don\u0027t have to keep\n \u003e re-voting each PS.\n \u003e \n \u003e Note, however, that part of my -2 was to signal \"don\u0027t bother\n \u003e spending a bunch of effort getting the design text ironed out,\n \u003e because this is unlikely to fly for non-design reasons\".\n\nThank Eric. I see you point. Personally I or my employer don\u0027t need this feature. I get behind it mostly because it seemed valuable for others in the room and I feel that we can contain the implementation impact pretty well. Anyhow I agree that we need a sort of consensus in the team to introduce this new top level concept to nova.","accounts_in_message":[],"_revision_number":13},{"id":"5ce26e4b961f5516cbd99e12193714fa1cf65af8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-11-20 10:10:16.000000000","message":"Patch Set 13:\n\n\u003e \u003e @Dan, @Eric: Do I understood your problems properly behind your\n \u003e \u003e -2s? Can we make steps forward with this or you still feel that\n \u003e \u003e this does not worth the effort?\n \u003e \n \u003e I actually don\u0027t want compose-able flavors at all, so I\u0027m at least\n \u003e -1 on that as a thing. I understand some do, and so I\u0027m definitely\n \u003e -2 on any proposal that moves to a world where things are only\n \u003e composed (as this spec was initially). IMHO, in order for something\n \u003e like this to proceed, we *must* have control over which flavors can\n \u003e be composed with which bits. Starting off with no restriction is\n \u003e not a better plan to me.\n\nOK. Let\u0027s drop the unconstrained idea of mine and try to come up with a solution where the admin can control which bit can be composed to which flavor. Like having a list of flavor ids for each bit and only allow composing the bit to the listed flavors.\n\n \u003e \n \u003e I think the underlying goal of this spec is to supposedly reduce\n \u003e the management overhead of defining flavors with various pieces of\n \u003e advanced extra specs, not just to provide the compose-able nature\n \u003e altogether. I challenge the assertion that this is a good solution\n \u003e to not having some scripts to automate that task and don\u0027t want to\n \u003e see us adopt something which I (and seemingly Eric) expect will\n \u003e bring significant complexity.\n\nSo you counter proposal is to automate flavor creation *on top of* the nova API. In theory such automation can implement composability in a same way giving possibility to the end user to choose from bits, compose them according to admin defined rules, and then this tool can create a flavor in nova as admin. \nThis counter proposal looks good to me too. \n\n@Brin, @Mohamed do you see this as a viable alternative to your use cases?\n\n \u003e \n \u003e If a deployer has a private cloud where everything should be\n \u003e compose-able by users (as this spec seemed to assert initially),\n \u003e why not just open up the policy to let users create flavors? Then\n \u003e they can create what they want, delete it them if/when they\u0027re\n \u003e finished, and use ansible or bash to template things out?\n\nI think this is a spectrum between use pure flavor and compose everything. Allowing flavor creation for end user is one extreme, while keeping flavor creation as admin only is the other. I can imagine that there is value behind letting the end user compose things but does not allow composing everything.\n\nAnyhow. Let\u0027s see what others think about your alternative proposal. \n\nThanks Dan.\n\ngibi","accounts_in_message":[],"_revision_number":13},{"id":"786c8e8128f09258a079bcc753530dbb4e76fad5","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-20 09:04:59.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"45a433c92e6e85e586c6f65c30735c9a727390f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 09:20:21.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9110b9d3292c4e67a06aa87a32870618 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5df145244246048c31f367557738ca : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":14},{"id":"8ae4b3e6cba8e72a61eb2cfbb175c35636afb002","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-01-20 01:01:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"29f2d27e1e2d486a4a80aa140953cad5f8d719d6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-22 00:30:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"029120d6d48d436b800cfaa2216b27028f1f79f6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-30 18:15:43.000000000","message":"Patch Set 14:\n\nCan we abandon this spec and close the blueprint?","accounts_in_message":[],"_revision_number":14},{"id":"1fdc3d1d2d2a1de92fc393e02dbbd80976267cd2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-05 06:54:34.000000000","message":"Patch Set 14:\n\n\u003e Can we abandon this spec and close the blueprint?\n\nSince we can\u0027t reach an agreement, we can abandon this SPEC, but this is not the only way to stop the flavor explosion. Perhaps some day in the future, more people will expect this change.","accounts_in_message":[],"_revision_number":14},{"id":"f92552502f00fcf4c02269b59ec30e6d0b0850b7","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-05 06:56:15.000000000","message":"Abandoned\n\nAbandoned because no consensus could be reached.","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"89e4d17191b518f5c638dacf14ed62f8c05abcfd","revisions":{"146ae95802c1d7268a43060ab39e45d9b651b603":{"kind":"REWORK","_number":1,"created":"2019-06-06 09:35:43.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/1"}}},"commit":{"parents":[{"commit":"37e4368601df54e4b01372b9919b65c807b3038c","subject":"Merge \"Amend count-quota-usage-from-placement to reflect implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/37e4368601df54e4b01372b9919b65c807b3038c"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:31:27.000000000","tz":480},"subject":"Add flavor metadata or metadata group","message":"Add flavor metadata or metadata group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nBlueprint: add-flavor-metadata-or-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/146ae95802c1d7268a43060ab39e45d9b651b603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/146ae95802c1d7268a43060ab39e45d9b651b603"}]},"branch":"refs/heads/master"},"da5507a7cb63797b13b56f9d97d67f9654012451":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-06 09:37:32.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/2"}}},"commit":{"parents":[{"commit":"37e4368601df54e4b01372b9919b65c807b3038c","subject":"Merge \"Amend count-quota-usage-from-placement to reflect implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/37e4368601df54e4b01372b9919b65c807b3038c"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:32:38.000000000","tz":480},"subject":"Add flavor metadata or metadata group","message":"Add flavor metadata or metadata group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-or-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da5507a7cb63797b13b56f9d97d67f9654012451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da5507a7cb63797b13b56f9d97d67f9654012451"}]},"branch":"refs/heads/master"},"71627db34a805575610b991268629331036fabdd":{"kind":"REWORK","_number":3,"created":"2019-06-27 10:17:09.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/3"}}},"commit":{"parents":[{"commit":"15da794b704f24e45fa0745243676e660cdee51f","subject":"grammar fix for show-server-numa-topology spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/15da794b704f24e45fa0745243676e660cdee51f"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-27 10:14:59.000000000","tz":480},"subject":"Add flavor metadata or metadata group","message":"Add flavor metadata or metadata group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-or-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/71627db34a805575610b991268629331036fabdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/71627db34a805575610b991268629331036fabdd"}]},"branch":"refs/heads/master"},"2b15430ded7bbd887b9c1481751d4f385d7584df":{"kind":"REWORK","_number":4,"created":"2019-06-28 10:36:44.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/4"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-28 10:34:52.000000000","tz":480},"subject":"Add flavor group","message":"Add flavor group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-or-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b15430ded7bbd887b9c1481751d4f385d7584df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b15430ded7bbd887b9c1481751d4f385d7584df"}]},"branch":"refs/heads/master"},"0ce0ff1fc6ae3151c6f55e9007c638dd74ba331f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-30 03:25:33.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/5"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-07-30 03:25:29.000000000","tz":0},"subject":"Add flavor group","message":"Add flavor group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-and-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ce0ff1fc6ae3151c6f55e9007c638dd74ba331f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ce0ff1fc6ae3151c6f55e9007c638dd74ba331f"}]},"branch":"refs/heads/master"},"b86045b9b2a94fb6b32a67b5f015535f864218fa":{"kind":"REWORK","_number":6,"created":"2019-10-28 11:16:41.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/6"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-10-28 11:16:37.000000000","tz":0},"subject":"Add flavor group","message":"Add flavor group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-and-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b86045b9b2a94fb6b32a67b5f015535f864218fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b86045b9b2a94fb6b32a67b5f015535f864218fa"}]},"branch":"refs/heads/master"},"4383b934f83126abf9975cfe69702252f859dd32":{"kind":"REWORK","_number":7,"created":"2019-11-03 15:30:15.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/7"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-11-03 15:25:38.000000000","tz":0},"subject":"Add flavor group","message":"Add flavor group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-and-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4383b934f83126abf9975cfe69702252f859dd32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4383b934f83126abf9975cfe69702252f859dd32"}]},"branch":"refs/heads/master"},"e8402a2e81ebb7d07653188756e66de691525a74":{"kind":"REWORK","_number":8,"created":"2019-11-04 09:06:27.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/8"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-11-04 09:02:06.000000000","tz":0},"subject":"Add flavor group","message":"Add flavor group\n\nThis blueprint proposes separate storage of extra specs that are not\nstrongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nCo-Authored-By: Guo Jingyu \u003cguojy8993@163.com\u003e\n\nBlueprint: add-flavor-metadata-and-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e8402a2e81ebb7d07653188756e66de691525a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e8402a2e81ebb7d07653188756e66de691525a74"}]},"branch":"refs/heads/master"},"cd159274e78a972b4f7117ae849afc3941741e7e":{"kind":"REWORK","_number":9,"created":"2019-11-10 15:47:45.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/9"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-11-10 15:47:37.000000000","tz":0},"subject":"Add flavor extra specs group","message":"Add flavor extra specs group\n\nThis blueprint proposes define a \"ComposableFlavorBit\" REST API entity that\ncan hold any kind of flavor bits (extra specs, normal flavor fields), and\nadd additional apis for flavor fragment management. Extend the server create\nREST API to allow the end user to specify the \"ComposableFlavorBit\"s he/she\nwants to add to the \"base\" flavor he/she used in the create request.This is\nnot strongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nBlueprint: add-flavor-metadata-and-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cd159274e78a972b4f7117ae849afc3941741e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cd159274e78a972b4f7117ae849afc3941741e7e"}]},"branch":"refs/heads/master"},"147f5c8dc5c4f68b923df941d0c4753d099f60c4":{"kind":"REWORK","_number":10,"created":"2019-11-14 08:55:58.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/10"}}},"commit":{"parents":[{"commit":"8d5540935ecdd2e2ea54313c0195ef08955a843a","subject":"Merge \"Enable rebuild for instances in cell0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d5540935ecdd2e2ea54313c0195ef08955a843a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-11-14 06:19:56.000000000","tz":0},"subject":"Add flavor extra specs group","message":"Add flavor extra specs group\n\nThis blueprint proposes define a \"ComposableFlavorBit\" REST API entity that\ncan hold any kind of flavor bits (extra specs, normal flavor fields), and\nadd additional apis for flavor fragment management. Extend the server create\nREST API to allow the end user to specify the \"ComposableFlavorBit\"s he/she\nwants to add to the \"base\" flavor he/she used in the create request.This is\nnot strongly associated with flavors, enhancing the reusability of flavors.\n\nAPIImpact\n\nBlueprint: add-flavor-metadata-and-metadata-group\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/147f5c8dc5c4f68b923df941d0c4753d099f60c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/147f5c8dc5c4f68b923df941d0c4753d099f60c4"}]},"branch":"refs/heads/master"},"564cc54aa15e10bd539b4eedd4e44ed34da99d37":{"kind":"REWORK","_number":11,"created":"2019-11-15 05:41:05.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/11"}}},"commit":{"parents":[{"commit":"acd5328f45f9dfd383de2b251fd1cc6a3f7b8701","subject":"Merge \"Add spec for VM-scoped SR-IOV NUMA affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/acd5328f45f9dfd383de2b251fd1cc6a3f7b8701"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-11-15 05:40:22.000000000","tz":480},"subject":"Add resources metadata of instance","message":"Add resources metadata of instance\n\nThis blueprint proposes define a \"ResourcesMetadata\" REST API entity\nthat can hold any kind of flavor bits (extra specs, normal flavor fields),\nand add additional apis for flavor fragment management. Then extend the\nserver create REST API to allow the end user to specify the\n\"ResourcesMetadata\"s they want to add to the base flavor they used in the\ncreate request.\n\nThis is not strongly associated with flavors, enhancing the reusability\nof flavors.\n\nAPIImpact\n\nBlueprint: resources-metadata-of-instance\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/564cc54aa15e10bd539b4eedd4e44ed34da99d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/564cc54aa15e10bd539b4eedd4e44ed34da99d37"}]},"branch":"refs/heads/master"},"24264aa44e98934a9499afcdb2145c3930f68a52":{"kind":"REWORK","_number":12,"created":"2019-11-15 06:45:33.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/12"}}},"commit":{"parents":[{"commit":"acd5328f45f9dfd383de2b251fd1cc6a3f7b8701","subject":"Merge \"Add spec for VM-scoped SR-IOV NUMA affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/acd5328f45f9dfd383de2b251fd1cc6a3f7b8701"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-11-15 06:44:51.000000000","tz":480},"subject":"Add resources metadata of instance","message":"Add resources metadata of instance\n\nThis blueprint proposes define a \"ResourcesMetadata\" REST API entity\nthat can hold any kind of flavor bits (extra specs, normal flavor fields),\nand add additional apis for flavor fragment management. Then extend the\nserver create REST API to allow the end user to specify the\n\"ResourcesMetadata\"s they want to add to the base flavor they used in the\ncreate request.\n\nThis is not strongly associated with flavors, enhancing the reusability\nof flavors.\n\nAPIImpact\n\nBlueprint: resources-metadata-of-instance\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/24264aa44e98934a9499afcdb2145c3930f68a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/24264aa44e98934a9499afcdb2145c3930f68a52"}]},"branch":"refs/heads/master"},"33f21ef706196527f8909178a7d4251abce2166e":{"kind":"REWORK","_number":13,"created":"2019-11-20 04:58:13.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/13"}}},"commit":{"parents":[{"commit":"b4c98982014b197b891200a3cb087b342522cbed","subject":"Merge \"Virtual instance rescue with stable disk devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b4c98982014b197b891200a3cb087b342522cbed"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-11-20 04:57:58.000000000","tz":480},"subject":"Add composable flavor properties","message":"Add composable flavor properties\n\nThis blueprint proposes define a \"ComposableFlavorProperties\" REST API\nentity that can hold any kind of extra specs, and add additional apis for\nflavor fragment management. Then extend the server create REST API to allow\nthe end user to specify the \"ComposableFlavorProperties\" they want to add to\nthe base flavor they used in the create request.\n\nThis is not strongly associated with flavors, enhancing the reusability of\nflavors.\n\nAPIImpact\n\nBlueprint: composable-flavor-properties\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33f21ef706196527f8909178a7d4251abce2166e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33f21ef706196527f8909178a7d4251abce2166e"}]},"branch":"refs/heads/master"},"89e4d17191b518f5c638dacf14ed62f8c05abcfd":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-12-20 09:04:59.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/663563/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/663563/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/663563/14"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-06 09:27:07.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-12-20 09:04:59.000000000","tz":0},"subject":"Add composable flavor properties","message":"Add composable flavor properties\n\nThis blueprint proposes define a \"ComposableFlavorProperties\" REST API\nentity that can hold any kind of extra specs, and add additional apis for\nflavor fragment management. Then extend the server create REST API to allow\nthe end user to specify the \"ComposableFlavorProperties\" they want to add to\nthe base flavor they used in the create request.\n\nThis is not strongly associated with flavors, enhancing the reusability of\nflavors.\n\nAPIImpact\n\nBlueprint: composable-flavor-properties\nChange-Id: Ied367a32d3785b29c4ba7b64a04c855c407a38e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/89e4d17191b518f5c638dacf14ed62f8c05abcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/89e4d17191b518f5c638dacf14ed62f8c05abcfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
