)]}'
{"id":"openstack%2Fnova-specs~664198","triplet_id":"openstack%2Fnova-specs~master~I8c0d949ecbdaaef3deb0829a9bfcd368db46132f","project":"openstack/nova-specs","branch":"master","topic":"bp/support-specifying-az-when-restore-shelved-server","hashtags":[],"change_id":"I8c0d949ecbdaaef3deb0829a9bfcd368db46132f","subject":"Clarify the error response in proposed change","status":"MERGED","created":"2019-06-10 06:29:20.000000000","updated":"2019-06-11 01:01:46.000000000","submitted":"2019-06-11 01:01:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664198-1560214906266-41fe81b9","meta_rev_id":"64fb4881aca567474a4ef48a81c149f489c63339","_number":664198,"virtual_id_number":664198,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-06-11 01:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":2,"date":"2019-06-10 15:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-06-11 00:48:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-06-11 00:48:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 15:12:26.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-11 00:48:41.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-06-11 01:01:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbddb87a5b0dee34b3b5f2c0c14081bbdba9f16e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-10 06:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1641bcce3924caeea4b4790efb4cf766b389032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 06:41:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/664198/1/check/openstack-tox-docs/36d01cd/html/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 http://logs.openstack.org/98/664198/1/check/openstack-tox-pep8/4aacf39/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2d47c4f2c11c0d6df95880b30b5f4cf552b64443","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-10 07:10:34.000000000","message":"Topic set to bp/support-specifying-az-when-restore-shelved-server","accounts_in_message":[],"_revision_number":1},{"id":"29d842071d3e0a1417b4b6b752743c561ba054f0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-10 15:12:26.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"6c07c58f99ee1adda6cebd01d0128582dc36cb1a","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-11 00:48:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0a21940d9b0f4d2bca74749ce08030baa3c1bfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:48:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a292f5756bf28c7de23a83943dc6f7a824650474","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:01:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"64fb4881aca567474a4ef48a81c149f489c63339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:01:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/664198/1/gate/openstack-tox-docs/d4c5a99/html/ : SUCCESS in 7m 51s\n- openstack-tox-pep8 http://logs.openstack.org/98/664198/1/gate/openstack-tox-pep8/28ab02d/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93904799878bef37eb26db463767da0cb657bbc6","revisions":{"93904799878bef37eb26db463767da0cb657bbc6":{"kind":"REWORK","_number":1,"created":"2019-06-10 06:29:20.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/98/664198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/98/664198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/98/664198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/98/664198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/98/664198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/98/664198/1"}}},"commit":{"parents":[{"commit":"014f3cfe7996567c101c5f74015e184adb2e0bb3","subject":"Specifying az when restore shelved server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/014f3cfe7996567c101c5f74015e184adb2e0bb3"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-10 06:28:17.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-10 06:28:21.000000000","tz":480},"subject":"Clarify the error response in proposed change","message":"Clarify the error response in proposed change\n\nRelated-on: 014f3cfe7996567c101c5f74015e184adb2e0bb3\nChange-Id: I8c0d949ecbdaaef3deb0829a9bfcd368db46132f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93904799878bef37eb26db463767da0cb657bbc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93904799878bef37eb26db463767da0cb657bbc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
