)]}'
{"id":"openstack%2Fnova-specs~666032","triplet_id":"openstack%2Fnova-specs~master~I468abe984d81c264a588f23d4b3804106339a597","project":"openstack/nova-specs","branch":"master","topic":"bp/cpu-resources","hashtags":[],"change_id":"I468abe984d81c264a588f23d4b3804106339a597","subject":"Additional upgrade clarifications for cpu-resources","status":"MERGED","created":"2019-06-18 15:52:34.000000000","updated":"2020-01-29 11:51:28.000000000","submitted":"2020-01-29 11:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666032-1580298574478-641ca554","meta_rev_id":"22f2bba8888a40e03b6ff1d22389cc462dfcdb82","_number":666032,"virtual_id_number":666032,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2020-01-29 11:46:03.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 11:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2020-01-29 11:15:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 11:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2020-01-29 10:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-29 11:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 22:10:06.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-08 04:57:05.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-10-23 00:12:20.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-01-29 10:32:27.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-01-29 11:15:15.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-01-29 11:38:15.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-29 11:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2247db20fd8fe2aa9539c1536cfc7fa60f28ca47","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-18 15:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"092dd530a75dada3d795c3e9c6991212857e8773","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:06:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/32/666032/1/check/openstack-tox-docs/e169496/html/ : SUCCESS in 7m 53s\n- openstack-tox-pep8 http://logs.openstack.org/32/666032/1/check/openstack-tox-pep8/e06f3d2/ : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0ccd77aae927fc401e85998e574b06f843537838","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-18 21:10:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f2a5d5dedf5b918dfc28f6245f64b64527f1b48","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-19 04:48:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"449c8179f789db98b6de9cac2600a4dc5cb74a32","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-19 04:50:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8615917c07a2611f04a3e73f146f28891a09060","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-19 15:28:31.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef7e235bdb86823a06f89d4df86d7e61a7162225","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-19 16:02:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"edcf9ab85cbaf1d81e0b6d88f32c265a0f312db9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:19:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/666032/2/check/openstack-tox-docs/a031016/html/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 http://logs.openstack.org/32/666032/2/check/openstack-tox-pep8/96253e5/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2353fcbe3ec1e2ccdb22f2c9a5c19d3838503320","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-07-04 12:32:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4922a70536bb11b38982a7b8036c5aedf3129a82","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 22:26:35.000000000","message":"Abandoned\n\nThis change proposes a spec for the Train release, which is currently under spec freeze. If you still wish your spec to be considered for Train, you may restore it and follow the spec freeze exception process per [1]. Alternatively, you may propose your spec to the backlog per [2], or wait until the U release opens and propose it there.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n[2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications","accounts_in_message":[],"_revision_number":2},{"id":"4dbcc6714925f8838a7002a947d8a9f34ff3dd8c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-30 09:32:59.000000000","message":"Restored\n\n\u003e Abandoned\n \u003e \n \u003e This change proposes a spec for the Train release, which is\n \u003e currently under spec freeze. If you still wish your spec to be\n \u003e considered for Train, you may restore it and follow the spec freeze\n \u003e exception process per [1]. Alternatively, you may propose your spec\n \u003e to the backlog per [2], or wait until the U release opens and\n \u003e propose it there.\n \u003e \n \u003e [1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n \u003e [2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications\n\nAmendment to an existing spec. This still makes sense","accounts_in_message":[],"_revision_number":2},{"id":"5527fffd41e1b089f5c8702e8ddc20e80bde6286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:42:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/666032/2/check/openstack-tox-docs/4f54895/html/ : SUCCESS in 8m 11s\n- openstack-tox-pep8 http://logs.openstack.org/32/666032/2/check/openstack-tox-pep8/0d4ba5d/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8ce008b55eb3924c18acfc1cc1502f2e281756bb","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-30 13:23:37.000000000","message":"Patch Set 2:\n\n\u003e Amendment to an existing spec. This still makes sense\n\nYup, sorry, my search picked up some stuff that shouldn\u0027t have been abandoned, my bad.","accounts_in_message":[],"_revision_number":2},{"id":"81981e1289391e008841b73441e1fe7bf8c5b754","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-23 22:10:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(4 comments)\n\nwfm.\n\nMight as well make these few edits, to avoid a fup to a fup.","accounts_in_message":[],"_revision_number":2},{"id":"376c940596352bc54741112f84e7a9a613d9eaaf","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-10-08 04:57:05.000000000","message":"Patch Set 2: -Code-Review\n\nThis is old one, I think, we need more update here?","accounts_in_message":[],"_revision_number":2},{"id":"d1c433f6f545a9ce0c5f8939a534c57ee3b73d0c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-10-23 00:12:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nShould be moved to specs/ussuri/approved.","accounts_in_message":[],"_revision_number":2},{"id":"5e34a5c90e185acfb2498ca20c2b1feb2ce039fb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 11:17:16.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"caacb0ae03696b0e71c6925cbf3c6adf92b650f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 11:17:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a5e56e9c2e1e32255ce288e631e80c9395beb80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 11:27:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313984315eb646fdb46a88d3eb8fee6f : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d394137fa7449b6a1783b36609ef901 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"628ae6f0dc9143f3fc695c7ff8c04fd3b584a907","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-29 10:32:27.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good and matches with the implementation.","accounts_in_message":[],"_revision_number":3},{"id":"803f510acc9a919436a567f647d386622c188b46","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 11:15:15.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nya i think this matches up to what i rememebr while i have made one nit inline i dont think it need to be respon to adress since this is a spec and not the offical docs for this. if this was a doc update i would prefer to see a more detailed explanation of how to configure a host that will be used for pinning.","accounts_in_message":[],"_revision_number":3},{"id":"8a51c19ec56d40b632ee17164a5f416a1e3e8ba1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-29 11:38:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"391bc494caa68d9cd4103d7c0b902a85c76db18e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:38:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"35b7dcf62251c524cf03bac7ffd53cb268e67ec9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 11:45:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d30fc830a730c65dfc5a501a16d13a2727f12fe9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 11:46:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c296c8a15dc060cf67b7af7cc80cca70002e5054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:49:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2430b3f6e5247ec846645999f578bf5 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79c3a028f2e4c209157b8f1fce54e3a : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"5c92a3c96007a21601846105821efd4037bd85cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:49:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"93d312ba11ec681af35762af78fda54ef0ea01c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:51:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/aef521d1ec5b42b2953f7988c1186900 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4b506d0e9f75200940aa41de419cc626b7dfbbbf","revisions":{"484b15a140a269d6073e3f1f67d5a42d51b0ba66":{"kind":"REWORK","_number":1,"created":"2019-06-18 15:52:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/666032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/32/666032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/1"}}},"commit":{"parents":[{"commit":"c8fdf0461b0bb1e710758ffbfe8b699d92bbf626","subject":"Update spec: filtering of alloc candidates by forbidden aggregates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c8fdf0461b0bb1e710758ffbfe8b699d92bbf626"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-06-18 15:52:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-06-18 15:52:09.000000000","tz":60},"subject":"Additional upgrade clarifications for cpu-resources","message":"Additional upgrade clarifications for cpu-resources\n\nBased on feedback from the mailing list.\n\nChange-Id: I468abe984d81c264a588f23d4b3804106339a597\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nBlueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/484b15a140a269d6073e3f1f67d5a42d51b0ba66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/484b15a140a269d6073e3f1f67d5a42d51b0ba66"}]},"branch":"refs/heads/master"},"0fc90beec57d5f0001ef2c85b4c414b4c41b28c9":{"kind":"REWORK","_number":2,"created":"2019-06-19 16:02:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/666032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/32/666032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/2"}}},"commit":{"parents":[{"commit":"c8fdf0461b0bb1e710758ffbfe8b699d92bbf626","subject":"Update spec: filtering of alloc candidates by forbidden aggregates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c8fdf0461b0bb1e710758ffbfe8b699d92bbf626"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-06-18 15:52:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-06-19 16:01:55.000000000","tz":60},"subject":"Additional upgrade clarifications for cpu-resources","message":"Additional upgrade clarifications for cpu-resources\n\nBased on feedback from the mailing list [1]. Key changes:\n\n- We don\u0027t require an operator set both \u0027cpu_shared_set\u0027 and\n  \u0027cpu_dedicated_set\u0027. Clarify why this is the case.\n\n- Add an upgrade summary and summaries to each upgrade step, since this\n  is by far the hairiest part of the whole exercise.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/thread.html#7084\n\nChange-Id: I468abe984d81c264a588f23d4b3804106339a597\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nBlueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fc90beec57d5f0001ef2c85b4c414b4c41b28c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fc90beec57d5f0001ef2c85b4c414b4c41b28c9"}]},"branch":"refs/heads/master"},"4b506d0e9f75200940aa41de419cc626b7dfbbbf":{"kind":"REWORK","_number":3,"created":"2020-01-27 11:17:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/666032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/32/666032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/32/666032/3"}}},"commit":{"parents":[{"commit":"4fd2bcd1892c6a121f821a47896e1e2d33f796ba","subject":"Merge \"Boot from volume instance rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4fd2bcd1892c6a121f821a47896e1e2d33f796ba"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-06-18 15:52:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-27 11:15:14.000000000","tz":0},"subject":"Additional upgrade clarifications for cpu-resources","message":"Additional upgrade clarifications for cpu-resources\n\nBased on feedback from the mailing list [1] and changes during\nimplementation. Key changes:\n\n- We don\u0027t require an operator set both \u0027cpu_shared_set\u0027 and\n  \u0027cpu_dedicated_set\u0027. Clarify why this is the case.\n\n- Add an upgrade summary and summaries to each upgrade step, since this\n  is by far the hairiest part of the whole exercise.\n\n- Replace references to an optional prefilter with the scheduler\n  aliasing with optional fallback functionality actually used\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/thread.html#7084\n\nChange-Id: I468abe984d81c264a588f23d4b3804106339a597\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nBlueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b506d0e9f75200940aa41de419cc626b7dfbbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4b506d0e9f75200940aa41de419cc626b7dfbbbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
