)]}'
{"id":"openstack%2Fnova-specs~667926","triplet_id":"openstack%2Fnova-specs~master~Ifc274ffd0cc2249cd49e77aea77a1e2b80ebd5cb","project":"openstack/nova-specs","branch":"master","topic":"bp/physical-tpm-passthrough","hashtags":[],"change_id":"Ifc274ffd0cc2249cd49e77aea77a1e2b80ebd5cb","subject":"WIP: Physical TPM passthrough","status":"ABANDONED","created":"2019-06-27 13:21:52.000000000","updated":"2019-07-24 14:40:40.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4adf9f6fe5c0916c99ae8c19fd45fb151f6738dc","_number":667926,"virtual_id_number":667926,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"date":"2019-06-27 21:27:45.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"date":"2019-06-27 19:24:03.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-06-27 13:32:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":-1,"date":"2019-06-28 22:23:03.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 13:32:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 19:24:03.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-27 21:27:45.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"075542ae49ad38ef1c370358b961cae013913bce","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 13:21:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1449ffec972fe9f5c156f70756ba3dae169f828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:32:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/26/667926/1/check/openstack-tox-docs/bec6560/html/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/26/667926/1/check/openstack-tox-pep8/9420895/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"44e28bbd81c0ec3a1e7183cf7deffee3d66c6de7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 13:57:40.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI don\u0027t think this is even close to consideration for Train and really think the people that want it (Oath) should be driving this if we\u0027re going to spend time reviewing this spec.","accounts_in_message":[],"_revision_number":1},{"id":"e0e85ba905097c27a578f2ac83cdc0e8c55b3751","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 14:44:18.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e I don\u0027t think this is even close to consideration for Train \n\nWhy\u0027s that?\n\n\u003e and really think the people that want it (Oath) should be driving this if we\u0027re going to spend time reviewing this spec.\n\nThey are; I\u0027m just helping out.","accounts_in_message":[],"_revision_number":1},{"id":"9806332f6106dba67c261647bd42bc2e5df8f75e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-06-27 19:05:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c33e83ada68741f3d56b11401072f2c24f22516","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 19:19:26.000000000","message":"Patch Set 1:\n\n\u003e\u003e I don\u0027t think this is even close to consideration for Train \n\u003e Why\u0027s that?\n\nMostly about the migration stuff - but if the requirement here is to actually never migrate these servers then it\u0027s less terrible than I originally thought. I just really don\u0027t like adding stuff to nova that doesn\u0027t work for most of the API (like boot from volume for a long time, and still is to an extent) and then no one is around to make it work later. BFV is a different animal and there are efforts to close those API feature parity gaps, but it\u0027s been a pain in the ass as you\u0027ve probably seen. I also wouldn\u0027t expect PTPM passthrough to be anywhere near the adoption from users as BFV so it\u0027s not really in the same ballpark, I\u0027m just trying to give an example of painful technical debt. GPUs/SRIOV/NUMA/SEV are also good examples and probably more relatable to this.","accounts_in_message":[],"_revision_number":1},{"id":"0c2d8a0574289dee27dfe255b3fcd0923e1ecf9a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 19:24:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"982f53ecdc4fa88273b615d05d67558ee5d45ffd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-06-27 21:27:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"71eceec5d307deeb2d18abe66bb43b1a495287d9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-28 22:23:03.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s take this off the radar for a minute. I\u0027m finding out things.","accounts_in_message":[],"_revision_number":1},{"id":"dc3f4a147cdd78a1847c67acd0d6f804485202aa","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-24 14:40:40.000000000","message":"Abandoned\n\nTurns out this is not workable. Among other reasons, clearing and transferring ownership of the pTPM requires a reboot.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae9fc6726a55db3058afc3c8f6c5c9e783773eb2","revisions":{"ae9fc6726a55db3058afc3c8f6c5c9e783773eb2":{"kind":"REWORK","_number":1,"created":"2019-06-27 13:21:52.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/667926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/26/667926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/667926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/667926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/667926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/26/667926/1"}}},"commit":{"parents":[{"commit":"15da794b704f24e45fa0745243676e660cdee51f","subject":"grammar fix for show-server-numa-topology spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/15da794b704f24e45fa0745243676e660cdee51f"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-26 23:10:52.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-27 13:21:10.000000000","tz":-300},"subject":"WIP: Physical TPM passthrough","message":"WIP: Physical TPM passthrough\n\nAs an alternative to virtual TPM [1], this spec describes passing the\nphysical TPM (Trusted Platform Module) device through to the guest.\nThough parts of the framework are usable by any hypervisor, only the\nlibvirt driver implementation is described.\n\n[1] http://specs.openstack.org/openstack/nova-specs/specs/train/approved/add-emulated-virtual-tpm.html\n\nChange-Id: Ifc274ffd0cc2249cd49e77aea77a1e2b80ebd5cb\nblueprint: physical-tpm-passthrough\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae9fc6726a55db3058afc3c8f6c5c9e783773eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae9fc6726a55db3058afc3c8f6c5c9e783773eb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
