)]}'
{"id":"openstack%2Fnova-specs~669970","triplet_id":"openstack%2Fnova-specs~master~I96472956ed6b0f75a0f299cfa5596ff9a8855a12","project":"openstack/nova-specs","branch":"master","topic":"bp/virtual-persistent-memory","hashtags":[],"change_id":"I96472956ed6b0f75a0f299cfa5596ff9a8855a12","subject":"update spec: virtual persistent memory","status":"ABANDONED","created":"2019-07-10 02:23:46.000000000","updated":"2019-07-29 22:26:33.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d0c9c1f9e3d7c05e152644191814c738c3eb7716","_number":669970,"virtual_id_number":669970,"owner":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":1,"date":"2019-07-10 02:35:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":-1,"date":"2019-07-10 08:53:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-07-12 22:50:39.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-07-12 22:50:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 02:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 08:53:19.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-07-12 22:50:33.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-07-12 22:50:39.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"05f1934c59a1d5ad63d7d4a93959560b24b7ee68","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-07-10 02:23:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca593eb3ea77900cf21a3028f7519a9d9898fac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 02:35:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/669970/1/check/openstack-tox-docs/84a09ac/html/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/70/669970/1/check/openstack-tox-pep8/8d635b3/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f83237e0425525f08d4e73f993b6e9203bd44dc6","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-07-10 08:53:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1dd0ef46977fa39def135f5bec3b7bab438b8ceb","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 22:26:33.000000000","message":"Abandoned\n\nThis change proposes a spec for the Train release, which is currently under spec freeze. If you still wish your spec to be considered for Train, you may restore it and follow the spec freeze exception process per [1]. Alternatively, you may propose your spec to the backlog per [2], or wait until the U release opens and propose it there.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n[2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bedb67ba9cd8e64472f7a8ce05625fc0912120b6","revisions":{"bedb67ba9cd8e64472f7a8ce05625fc0912120b6":{"kind":"REWORK","_number":1,"created":"2019-07-10 02:23:46.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/70/669970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/70/669970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/70/669970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/70/669970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/70/669970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/70/669970/1"}}},"commit":{"parents":[{"commit":"42ba77612edefafa5c366a759f003e70ee8090c5","subject":"Merge \"Fix the cross_az_attach value in AZ-unshelved spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/42ba77612edefafa5c366a759f003e70ee8090c5"}]}],"author":{"name":"Luyao Zhong","email":"luyao.zhong@intel.com","date":"2019-07-10 02:19:15.000000000","tz":480},"committer":{"name":"Luyao Zhong","email":"luyao.zhong@intel.com","date":"2019-07-10 02:27:47.000000000","tz":480},"subject":"update spec: virtual persistent memory","message":"update spec: virtual persistent memory\n\nUpdate VirtualPMEM object to support pmem data migration,\notherwise we can\u0027t get pmem info from remote host. And\nuse \u0027label\u0027 instead of \u0027rc_name\u0027, because ‘rc_name’ is\na little confusing, actually only the \u0027label\u0027 name will\nbe saved but not the resource class name.\n\nChange-Id: I96472956ed6b0f75a0f299cfa5596ff9a8855a12\nImplements: blueprint virtual-persistent-memory\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bedb67ba9cd8e64472f7a8ce05625fc0912120b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bedb67ba9cd8e64472f7a8ce05625fc0912120b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
