)]}'
{"id":"openstack%2Fnova-specs~672400","triplet_id":"openstack%2Fnova-specs~master~I0f737297dcb1c0cb7eed1f03e544e708b8762981","project":"openstack/nova-specs","branch":"master","topic":"bp/boot-with-vnic-type","hashtags":[],"change_id":"I0f737297dcb1c0cb7eed1f03e544e708b8762981","subject":"Support set port vnic type to create server","status":"ABANDONED","created":"2019-07-24 03:09:41.000000000","updated":"2020-03-09 03:00:59.000000000","total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6eb6305edc77eee33da8999afb324c922db6b536","_number":672400,"virtual_id_number":672400,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-12 13:09:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-2,"date":"2020-02-12 13:31:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 05:46:17.000000000","updated_by":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"reviewer":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"state":"REVIEWER"},{"updated":"2019-07-24 06:21:28.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-07-24 06:51:48.000000000","updated_by":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"reviewer":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"state":"REVIEWER"},{"updated":"2019-07-25 01:42:42.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-09-04 15:48:09.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-04 22:22:49.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2020-01-27 17:09:33.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-30 21:46:57.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-02-12 12:54:59.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-12 13:09:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-12 13:31:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d863778979959053f199e593a38fed5a1b8957ff","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-07-24 03:09:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"807b228be9a479049a94ae17da443f2a9b092ab7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 03:16:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/672400/1/check/openstack-tox-docs/860164f/html/ : SUCCESS in 5m 45s\n- openstack-tox-pep8 http://logs.openstack.org/00/672400/1/check/openstack-tox-pep8/bf43a79/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ce960ab71ac761b08d420943c47aa21d12df11d4","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-07-24 03:26:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"544b9d434271622e8cfd3a3fe596d72618c4196c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 03:34:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/672400/2/check/openstack-tox-docs/42b6832/html/ : SUCCESS in 7m 02s\n- openstack-tox-pep8 http://logs.openstack.org/00/672400/2/check/openstack-tox-pep8/e1ed5d2/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"76e9f0e082f6733202047eb7feccb00f8a6bdfe5","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-07-24 06:21:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"41d691ca20cec2dd294240e537eb7e7e5cfb8da3","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2019-07-24 06:51:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d70fe33b8596219caa455ccb9326f8dee22636e","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-07-24 07:03:32.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4366cc642a8137f52ace594cb0f27c28823eb7d8","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-07-24 08:28:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"444f7945b54ef35a0241fdaf0cb69bbda9675f4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 08:36:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/672400/3/check/openstack-tox-docs/afbfde1/html/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 http://logs.openstack.org/00/672400/3/check/openstack-tox-pep8/51b452b/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1bb8c301f5d696b41c9dc9cc7309e399bba7e95a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-25 01:51:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2da7878763f8f0507ca9e0957ba2e6c468f7dd5","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-25 03:48:07.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks Matt.","accounts_in_message":[],"_revision_number":3},{"id":"42b02cb9e8ef9c450e6c59c4f78be795eba09422","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 22:26:33.000000000","message":"Abandoned\n\nThis change proposes a spec for the Train release, which is currently under spec freeze. If you still wish your spec to be considered for Train, you may restore it and follow the spec freeze exception process per [1]. Alternatively, you may propose your spec to the backlog per [2], or wait until the U release opens and propose it there.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n[2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications","accounts_in_message":[],"_revision_number":3},{"id":"43a07c0f5f5403a77adb327f1f9c9374f98a4e26","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-30 13:54:04.000000000","message":"Patch Set 3:\n\nThe nova-specs repository has been populated for ussuri. You may restore and move this spec if needed.","accounts_in_message":[],"_revision_number":3},{"id":"4ac12cd1fc3f1257e8c3057a2f9694ee141d5cb8","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-09-02 07:17:07.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"ea243112286b606dc513dc14aeb73157dab7265b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 07:34:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891af371ae5149e39b94129360c264f2 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7aed810fe54204b7ed57201248a3c9 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e031515a74d3c97f570c0cc26cf19da0ee79e1d6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-02 10:20:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e51847fa94c04669717f4354d9064368b6db7ad4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-02 10:25:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ab2c8c4ae23743fcb3d980cb021b4ec88ca9a322","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 10:36:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ffabf175b4b4c0989f91137679bb2cb : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f199056bde410fb069c473e275d36c : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e20598b3c1c98d0d88f3a61305454d6e507a80cf","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-04 15:48:09.000000000","message":"Patch Set 5:\n\nFollowing on from Matt\u0027s comment, I feel like this potentially opens the floodgates to being able to specify all possible port information in a nova boot request.\n\nThat said, how is it different from the concept of device_profiles we\u0027re adding with cyborg integration [1]? In particular, the intended future direction of being able to pass them into the API independently of the flavor.\n\n[1] http://specs.openstack.org/openstack/nova-specs/specs/train/approved/nova-cyborg-interaction.html#user-requests","accounts_in_message":[],"_revision_number":5},{"id":"9e23c2a6ea8bd7aa0f6e0debd9bbcc27e2abadcd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 22:03:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"037081fd5ae4961db17c9a6ce4d622d3b4ee3dcf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 22:06:03.000000000","message":"Patch Set 5: Code-Review-1\n\nI\u0027m still a -1 on this for the same reason I said before [1]. \"We have some customers asking for this\" is not sufficient justification for this in my opinion, especially when there alternatives available. The frustration I\u0027ve had at Huawei with this type of customer-first culture is that it doesn\u0027t seem a lot of thought is put into focusing the effort on higher level tooling (orchestrators, CLIs, UIs) which the user would use rather than directly shoving this type of complexity into the low level compute APIs themselves.\n\n[1] https://review.opendev.org/#/c/672400/3/specs/train/approved/boot-with-vnic-type.rst@43","accounts_in_message":[],"_revision_number":5},{"id":"579b10d3cfb20145baa1d0824c7658ff4f4766db","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 22:12:23.000000000","message":"Patch Set 5: -Code-Review\n\nI guess I didn\u0027t -1 before, so I\u0027ll drop the -1 but I\u0027m not really in support of this either. Maybe others are.","accounts_in_message":[],"_revision_number":5},{"id":"b173fd0ab68739fcd367abd703564e59df21c496","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 22:21:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0d57b5c54fc632990638cd1ec02dd2968aef34c9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 22:22:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"de21590b00b108522fe2a7e79c945b6def36de3a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-05 03:15:19.000000000","message":"Patch Set 5:\n\n\u003e I\u0027m still a -1 on this for the same reason I said before [1]. \"We\n \u003e have some customers asking for this\" is not sufficient\n \u003e justification for this in my opinion, especially when there\n \u003e alternatives available. The frustration I\u0027ve had at Huawei with\n \u003e this type of customer-first culture is that it doesn\u0027t seem a lot\n \u003e of thought is put into focusing the effort on higher level tooling\n \u003e (orchestrators, CLIs, UIs) which the user would use rather than\n \u003e directly shoving this type of complexity into the low level compute\n \u003e APIs themselves.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/672400/3/specs/train/approved/boot-with-vnic-type.rst@43\n\nI think I can understand what you are thinking about. Perhaps, provide a layer with high leavel tools (orchestrators, CLIs, UIs) for customer that handles that orchestration can perfect resolve this spec is pointing.\nI think you rarely go to the customer site or use a UI interface with multiple combinations, which is bad. When the request effect is achieved in the UI by combining the APIs of multiple components, this will undoubtedly increase the corresponding time of the UI request and make the UI become abnormally slow. This is also a problem that customers often complain about.","accounts_in_message":[],"_revision_number":5},{"id":"3fe54b6d2879f13a30ab6d04f48ed27fa8280020","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-27 17:09:33.000000000","message":"Patch Set 5: Code-Review-1\n\nI think Heat can be used to automate the two step creation (first port, then server with that port). I would not add more complexity to nova to something that can already be done. It might be unfortunate that OpenStack does not have one single API but I think it is what it is now.","accounts_in_message":[],"_revision_number":5},{"id":"cd8ed770e86fb33de088f371d0e9c68163e4b911","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-30 21:46:57.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)\n\ni have given up on makeing this change at this point.\ni have brought this up at the ptg/desgin summit at least 3 times\nso i think this is a valid customer ask but i dont think its in line with https://github.com/openstack/nova/blob/master/doc/source/contributor/project-scope.rst#no-more-api-proxies or https://github.com/openstack/nova/blob/master/doc/source/contributor/project-scope.rst#no-more-orchestration\n\nit does come up enough that we might want to consider it due to the ux improvement but this could be done client side too.\ni.e. osc could add this functionality without extending the nova api by just createing the ports in neutron then calling nova with the port id.","accounts_in_message":[],"_revision_number":5},{"id":"a7aac87db8107a1efea6ca6354c42e3414f44600","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-10 13:49:25.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThanks gibi, sean. We will update it soon.","accounts_in_message":[],"_revision_number":5},{"id":"41862b4f69d504eefb18bbd6210004efe58924ab","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2020-02-12 09:30:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2a89e5031ba00a0acd95890cd87c7899d12bf09b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 09:46:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be22c01c04b2486bb3e44f079524e4ca : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6141e3d23444ffa7f50f71e5922f6d : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":6},{"id":"9c3516857bd8d4f67eb8b2b3e89dd0766f9da0da","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2020-02-12 11:50:05.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0133eb5f45ed172ec95758755ccc9602c27571f7","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2020-02-12 11:51:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"30a6fbc24763c0789b2c63b3f126ab632c58e1e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:05:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b62e7a4f93c34651a7d7cd79d27b2b4a : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7c1dc0d1b9491d931283ec5c0dbd4f : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":7},{"id":"5280f173937c194bc297d7a45ee12f38d62f0a35","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2020-02-12 12:54:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"368ac4c9973ee1b16177148cd9eb1509cdc9cc96","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-12 12:54:59.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nThe pep8 ERROR that you were lost Feature Liaison section.","accounts_in_message":[],"_revision_number":7},{"id":"a131020163fdbde7edc54149ad7a20b2933ed81d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 13:09:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a933837cbecb4c02a88bde3eb6a706f8 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6056a2a62c9444f3b127458bb520f1b4 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":8},{"id":"6c3be1251e5112da3f240e8f06858c7a0e9ad2b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-12 13:31:13.000000000","message":"Patch Set 8: Code-Review-2\n\nI understand the reason you want this, but as mriedem, gibi, sean-k-mooney and more have noted, orchestration is not something we want nova to do anymore. There\u0027s no reason you couldn\u0027t provide custom tooling, via an OSC plugin or something else, to do this, but it doesn\u0027t belong in nova core.","accounts_in_message":[],"_revision_number":8},{"id":"c71d580142adb972f8ce82b7b3604a1804f4fe6f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 17:55:57.000000000","message":"Abandoned\n\nWe\u0027re at spec freeze and this has a -2.\n\nIf you wish to continue the discussion for a subsequent $release, you may:\n- Restore (or ask a nova-core to do so)\n- Move the spec from specs/ussuri/ to specs/$release/\n- Update the blueprint\u0027s Series Goal","accounts_in_message":[],"_revision_number":8},{"id":"6b87fec195a787fbfd4a292ebcbb4e125790d0ae","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-03-09 03:00:59.000000000","message":"Patch Set 8:\n\n\u003e I understand the reason you want this, but as mriedem, gibi,\n \u003e sean-k-mooney and more have noted, orchestration is not something\n \u003e we want nova to do anymore. There\u0027s no reason you couldn\u0027t provide\n \u003e custom tooling, via an OSC plugin or something else, to do this,\n \u003e but it doesn\u0027t belong in nova core.\n\nOk. More than half of Core does not want to do this function in Nova, I do not insist on continuing, as you said, through Shell, OSC we can still achieve these functions.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"eb81cfeb9d589d774bd4c79be3b76e99f3a8a7cc","revisions":{"2103ba70e81b36b4bd537bcb6e28e5663f183304":{"kind":"REWORK","_number":1,"created":"2019-07-24 03:09:41.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/00/672400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/1"}}},"commit":{"parents":[{"commit":"c50c3e6412e4d6c0bea02737ad6dfc131b58907f","subject":"Spec: Provider config YAML file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c50c3e6412e4d6c0bea02737ad6dfc131b58907f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:09:37.000000000","tz":480},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\nBlueprint: boot-with-vnic-type\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2103ba70e81b36b4bd537bcb6e28e5663f183304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2103ba70e81b36b4bd537bcb6e28e5663f183304"}]},"branch":"refs/heads/master"},"66d8258bbfe20b0c13cc50d04aa8f153d3da767b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-24 03:26:19.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/00/672400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/2"}}},"commit":{"parents":[{"commit":"c50c3e6412e4d6c0bea02737ad6dfc131b58907f","subject":"Spec: Provider config YAML file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c50c3e6412e4d6c0bea02737ad6dfc131b58907f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:25:55.000000000","tz":0},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/66d8258bbfe20b0c13cc50d04aa8f153d3da767b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/66d8258bbfe20b0c13cc50d04aa8f153d3da767b"}]},"branch":"refs/heads/master"},"e2fd017887132a125566c777019153683c95b6b1":{"kind":"REWORK","_number":3,"created":"2019-07-24 08:28:07.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/00/672400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/3"}}},"commit":{"parents":[{"commit":"c50c3e6412e4d6c0bea02737ad6dfc131b58907f","subject":"Spec: Provider config YAML file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c50c3e6412e4d6c0bea02737ad6dfc131b58907f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 08:27:15.000000000","tz":480},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e2fd017887132a125566c777019153683c95b6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e2fd017887132a125566c777019153683c95b6b1"}]},"branch":"refs/heads/master"},"64f814e7f2612c09290c861916b8215eaca53b18":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-02 10:20:04.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/00/672400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/4"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-09-02 10:20:04.000000000","tz":0},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64f814e7f2612c09290c861916b8215eaca53b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64f814e7f2612c09290c861916b8215eaca53b18"}]},"branch":"refs/heads/master"},"a12a664af839be19fd2b3741500c4499c045d8b8":{"kind":"REWORK","_number":5,"created":"2019-09-02 10:25:04.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/00/672400/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/5"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-02 10:19:31.000000000","tz":480},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a12a664af839be19fd2b3741500c4499c045d8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a12a664af839be19fd2b3741500c4499c045d8b8"}]},"branch":"refs/heads/master"},"1707de91197fd50a96b4d7af357283435e609775":{"kind":"REWORK","_number":6,"created":"2020-02-12 09:30:22.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/00/672400/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/6"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2020-02-12 09:24:32.000000000","tz":480},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1707de91197fd50a96b4d7af357283435e609775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1707de91197fd50a96b4d7af357283435e609775"}]},"branch":"refs/heads/master"},"215a86a427d48a6e3ab2819b06529555e701f454":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-02-12 11:51:45.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/00/672400/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/7"}}},"commit":{"parents":[{"commit":"4aefa3596eaafcae56ffa446b761ce29886b1695","subject":"Merge \"Support multiple store of Glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4aefa3596eaafcae56ffa446b761ce29886b1695"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2020-02-12 11:51:45.000000000","tz":0},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/215a86a427d48a6e3ab2819b06529555e701f454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/215a86a427d48a6e3ab2819b06529555e701f454"}]},"branch":"refs/heads/master"},"eb81cfeb9d589d774bd4c79be3b76e99f3a8a7cc":{"kind":"REWORK","_number":8,"created":"2020-02-12 12:54:13.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/00/672400/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/672400/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/672400/8"}}},"commit":{"parents":[{"commit":"4aefa3596eaafcae56ffa446b761ce29886b1695","subject":"Merge \"Support multiple store of Glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4aefa3596eaafcae56ffa446b761ce29886b1695"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-07-24 03:06:20.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2020-02-12 12:12:09.000000000","tz":480},"subject":"Support set port vnic type to create server","message":"Support set port vnic type to create server\n\nWhen users create server use network uuid to provision the server instance\nwith a NIC. The type of vNIC default value is ``normal``, user can not set\nit. This blueprint proposed to set port ``vnic_type`` when create a server.\n\nAPIImpact\n\nCo-Authored-By: Brin Zhang \u003czhangbailin@inspur.com\u003e\n\nBlueprint: boot-with-vnic-type\nChange-Id: I0f737297dcb1c0cb7eed1f03e544e708b8762981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb81cfeb9d589d774bd4c79be3b76e99f3a8a7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb81cfeb9d589d774bd4c79be3b76e99f3a8a7cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
