)]}'
{"id":"openstack%2Fnova-specs~673356","triplet_id":"openstack%2Fnova-specs~master~I32223c10cd86379b4d5c337e31d36f5d7820459e","project":"openstack/nova-specs","branch":"master","topic":"fast-specs","hashtags":[],"change_id":"I32223c10cd86379b4d5c337e31d36f5d7820459e","subject":"tox -e fast-specs","status":"MERGED","created":"2019-07-29 19:35:02.000000000","updated":"2019-10-08 14:36:25.000000000","submitted":"2019-10-08 14:33:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673356-1570545190017-fe29816b","meta_rev_id":"ece64402ae67d040ea1365a8cfd7794cea249a22","_number":673356,"virtual_id_number":673356,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 14:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2019-08-22 06:05:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-08-19 18:43:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-10-08 14:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-21 18:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-10-08 14:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 13:17:11.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2019-08-19 18:43:02.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-08-21 18:57:17.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-08-22 06:05:05.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-10-08 14:16:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-08 14:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d446f5926c0d1b9c3603bc54f99083026deb4e00","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 19:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57d168ea257a741a58b0524ac88f10f99afe78b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/673356/1/check/openstack-tox-docs/a744546/html/ : SUCCESS in 7m 09s\n- openstack-tox-pep8 http://logs.openstack.org/56/673356/1/check/openstack-tox-pep8/89b6b12/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"79c5cda7839e8b36cc5a6880c66335d8a463f6bf","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-08-19 13:17:11.000000000","message":"Patch Set 1: Code-Review+1\n\nI\u0027d give a +100 to this if Gerrit allowed me.","accounts_in_message":[],"_revision_number":1},{"id":"f28f24cfdf9f88fdd6481c76eb74c6b0676caeae","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-19 15:07:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)\n\ni think the logic makes sense but just because\nit bash does not mean we should not use good function and variable names","accounts_in_message":[],"_revision_number":1},{"id":"a358913e15589875fafcee4c4f24ba38a3ab0d41","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-19 15:35:42.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nI\u0027m OK with this since playing with it locally. I hope to never have to debug it, but since it\u0027s an opt-in personal little helper I\u0027m OK with it as-is. Will +2 now if Sean is OK with addressing his comments in a follow up, or just ping me once you\u0027ve addressed his comments and I\u0027ll +2 then.","accounts_in_message":[],"_revision_number":1},{"id":"3f1e94fe2deed39864169ac25448be5db594576c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 15:37:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22d97eb292370e8131189c7c387807fe21b86286","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 15:38:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"46bb6ab08ddb8d80657a2ee1adeff84517500c6b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 15:39:12.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fd7707e92f5b73f61ab9a9e577dc7fdc584c9c18","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-19 15:56:57.000000000","message":"Patch Set 3: Code-Review+2\n\nStill looks good to me. Pulled down PS3 and played around and it\u0027s working as expected.","accounts_in_message":[],"_revision_number":3},{"id":"835c35769263a1b2a3d9ff456f0bcc1a71219743","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-19 16:04:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a9b7133acec1785ea428b4ac809d2b0fe74911ed","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-19 16:06:48.000000000","message":"Patch Set 3: Code-Review+1\n\ni would still prefer to fix the other shellcheck error but this is much more readable so im ok with this version.","accounts_in_message":[],"_revision_number":3},{"id":"5b8124af7ee102bfb8621382e77da0f11ee7697f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:18:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013a9c9fedac443dababd9cdf60f39b9 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100d3ef5994f4e7a9b0a7967b76965e6 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"553ae377eb8aecc8f75107ab03ec050f3d4fe070","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 16:37:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d5b793da3e13dffea1d7c8596a292a100a4c2809","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 16:37:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2585b82a779de22685b57e574e1db7f4980dbdbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:53:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b1b0dac5b841649858911fc1c2cd5e : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e904c8409bd54986a42d9000124bff8e : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"9280e3eecf83521674147a2ca713b1a77a75c44f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-19 18:43:02.000000000","message":"Patch Set 4: Code-Review+1\n\nim still happy with this so +1\n\negrep has caused issues for some on windows with cygwin or osx i belive in the past as i dont think its installed by default but it is available.\n\nmost systems would proably be fine but grep is still alot more common.","accounts_in_message":[],"_revision_number":4},{"id":"e52d138babf12d3cddbdf3c1c62cf10db9d3e8bc","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-21 18:57:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0af8a0fd019bd002a0d268c358e0de3ca67c856d","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-22 06:05:05.000000000","message":"Patch Set 4: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":4},{"id":"1c93ce659978d6fbf798805da1018bceafcada67","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-08 14:16:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\neh, why not","accounts_in_message":[],"_revision_number":4},{"id":"bc2fa50522a1c83b363a2ea2b7703e36cea91471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:16:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"63c15125bc713cd47fea996ba0fcac74a56d74f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:33:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a6946597a642bd891f0de0863306c2 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d908265c7b7e4ad5b4924ceb8478b7f9 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"2d3370e6ca07dc8d012d8428a82b39c5a3af33af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:33:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"de5b97f2e58176979d43dc53537133f354174188","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:36:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/b6085999751047eeb5dc14076f33aadd : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4990806a8ddff3b3ea241ff0e36aa5942fb6dc93","revisions":{"c26dc44d7a5cbd0d0844c9af4430b24377d45415":{"kind":"REWORK","_number":1,"created":"2019-07-29 19:35:02.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/56/673356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/56/673356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/1"}}},"commit":{"parents":[{"commit":"4498387731c313c2a790396b40d758deb85bf284","subject":"Merge \"Add user_id to the migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4498387731c313c2a790396b40d758deb85bf284"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-29 19:21:38.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-29 19:35:00.000000000","tz":-300},"subject":"tox -e fast-specs","message":"tox -e fast-specs\n\nAdd a utility, tools/fast-specs.sh, and a corresponding tox target,\nfast-specs, to build only spec files changed since the last commit. This\nis way faster than building all the specs with tox -e docs, and saves\nyou having to know several quirks of sphinx-build to do it yourself.\n\nChange-Id: I32223c10cd86379b4d5c337e31d36f5d7820459e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c26dc44d7a5cbd0d0844c9af4430b24377d45415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c26dc44d7a5cbd0d0844c9af4430b24377d45415"}]},"branch":"refs/heads/master"},"fd21856be6312458d690288c2b545e093d2b5956":{"kind":"REWORK","_number":2,"created":"2019-08-19 15:37:46.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/56/673356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/56/673356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/2"}}},"commit":{"parents":[{"commit":"4498387731c313c2a790396b40d758deb85bf284","subject":"Merge \"Add user_id to the migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4498387731c313c2a790396b40d758deb85bf284"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-29 19:21:38.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-19 15:37:44.000000000","tz":-300},"subject":"tox -e fast-specs","message":"tox -e fast-specs\n\nAdd a utility, tools/fast-specs.sh, and a corresponding tox target,\nfast-specs, to build only spec files changed since the last commit. This\nis way faster than building all the specs with tox -e docs, and saves\nyou having to know several quirks of sphinx-build to do it yourself.\n\nChange-Id: I32223c10cd86379b4d5c337e31d36f5d7820459e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fd21856be6312458d690288c2b545e093d2b5956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fd21856be6312458d690288c2b545e093d2b5956"}]},"branch":"refs/heads/master"},"3bbbc5b74f900c3c3ee31068655a869d08fa8cf6":{"kind":"REWORK","_number":3,"created":"2019-08-19 15:38:54.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/56/673356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/56/673356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/3"}}},"commit":{"parents":[{"commit":"4498387731c313c2a790396b40d758deb85bf284","subject":"Merge \"Add user_id to the migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4498387731c313c2a790396b40d758deb85bf284"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-29 19:21:38.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-19 15:38:52.000000000","tz":-300},"subject":"tox -e fast-specs","message":"tox -e fast-specs\n\nAdd a utility, tools/fast-specs.sh, and a corresponding tox target,\nfast-specs, to build only spec files changed since the last commit. This\nis way faster than building all the specs with tox -e docs, and saves\nyou having to know several quirks of sphinx-build to do it yourself.\n\nChange-Id: I32223c10cd86379b4d5c337e31d36f5d7820459e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3bbbc5b74f900c3c3ee31068655a869d08fa8cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3bbbc5b74f900c3c3ee31068655a869d08fa8cf6"}]},"branch":"refs/heads/master"},"4990806a8ddff3b3ea241ff0e36aa5942fb6dc93":{"kind":"REWORK","_number":4,"created":"2019-08-19 16:37:20.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/56/673356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/56/673356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/56/673356/4"}}},"commit":{"parents":[{"commit":"4498387731c313c2a790396b40d758deb85bf284","subject":"Merge \"Add user_id to the migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4498387731c313c2a790396b40d758deb85bf284"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-29 19:21:38.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-19 16:37:18.000000000","tz":-300},"subject":"tox -e fast-specs","message":"tox -e fast-specs\n\nAdd a utility, tools/fast-specs.sh, and a corresponding tox target,\nfast-specs, to build only spec files changed since the last commit. This\nis way faster than building all the specs with tox -e docs, and saves\nyou having to know several quirks of sphinx-build to do it yourself.\n\nChange-Id: I32223c10cd86379b4d5c337e31d36f5d7820459e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4990806a8ddff3b3ea241ff0e36aa5942fb6dc93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4990806a8ddff3b3ea241ff0e36aa5942fb6dc93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
