)]}'
{"id":"openstack%2Fnova-specs~675024","triplet_id":"openstack%2Fnova-specs~master~I16977b3c14478e55240501f4b32f029909c9e25a","project":"openstack/nova-specs","branch":"master","topic":"update_vpmem","hashtags":[],"change_id":"I16977b3c14478e55240501f4b32f029909c9e25a","subject":"Update the virtual persistent memory spec based on discussion","status":"ABANDONED","created":"2019-08-07 04:15:35.000000000","updated":"2019-08-22 02:39:40.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a15da6fa6d47a76db15e5ec0f7f84aaba3d283f2","_number":675024,"virtual_id_number":675024,"owner":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-08-07 04:25:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":1,"date":"2019-08-14 01:54:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":-1,"date":"2019-08-20 04:49:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 04:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 01:54:28.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"242d8c11d3c9300aa83c25b0853f7220006ae362","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-08-07 04:15:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a286b4987e3b0802c54ce2197ed88edccdf93cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 04:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/24/675024/1/check/openstack-tox-docs/9f41cd8/html/ : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://logs.opendev.org/24/675024/1/check/openstack-tox-pep8/4eae633/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"411f78e2ba399b3af55934f0851cee071b730ada","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-08-07 07:11:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f7964e0c5ac9d77a5ff9e59f2e3531f8dad3f0a1","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-08-12 07:16:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7598f4b7798bd8bb07e6927c9b44e5cbff4ab7da","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-08-14 01:54:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"bbd7e2659e9d20752bebb9542688a944a74f8d2f","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-08-20 04:49:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a5b2978342a485b3e66486021bd08784221f9c14","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-08-22 02:39:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e94d53c54cf2fed9dc400b7dccbd8222826ebca","revisions":{"2e94d53c54cf2fed9dc400b7dccbd8222826ebca":{"kind":"REWORK","_number":1,"created":"2019-08-07 04:15:35.000000000","uploader":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"ref":"refs/changes/24/675024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/675024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/675024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/675024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/675024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/675024/1"}}},"commit":{"parents":[{"commit":"e3d3caef31e0028a89bbdc4766dea78058bef596","subject":"Fix URL to blueprint for Secure Boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3d3caef31e0028a89bbdc4766dea78058bef596"}]}],"author":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-08-07 04:21:37.000000000","tz":480},"committer":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-08-07 04:21:37.000000000","tz":480},"subject":"Update the virtual persistent memory spec based on discussion","message":"Update the virtual persistent memory spec based on discussion\n\nUpdate the spec based on the IRC discussion:\n* http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-07-15.log.html#t2019-07-15T15:03:08\n* http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-07-17.log.html#t2019-07-17T15:28:46\n\nChange-Id: I16977b3c14478e55240501f4b32f029909c9e25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2e94d53c54cf2fed9dc400b7dccbd8222826ebca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2e94d53c54cf2fed9dc400b7dccbd8222826ebca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
