)]}'
{"id":"openstack%2Fnova-specs~675276","triplet_id":"openstack%2Fnova-specs~master~Iaff9b643bdfafb9ec26ac430bebb1a094816ef85","project":"openstack/nova-specs","branch":"master","topic":"bp/cpu-resources","hashtags":[],"change_id":"Iaff9b643bdfafb9ec26ac430bebb1a094816ef85","subject":"Correct nit with cpu-resources","status":"MERGED","created":"2019-08-08 09:13:25.000000000","updated":"2019-08-09 09:57:48.000000000","submitted":"2019-08-09 09:57:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675276-1565344668880-c4f9ddb8","meta_rev_id":"f4857971afa86a2903040162166635bb3f1ad785","_number":675276,"virtual_id_number":675276,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-08-09 09:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-08-09 09:29:57.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-08-09 09:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 14:57:59.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-09 09:57:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c9955f358bf7b28055babd98a2eb3e118ed8dce","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-08 09:13:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da42ec9edd941da21a66e5151424ccd710f6748a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:31:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/76/675276/1/check/openstack-tox-docs/8497613/html/ : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://logs.opendev.org/76/675276/1/check/openstack-tox-pep8/f8a267e/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f3e3b3799d54aef42eb72eba8df4fc8c2c144460","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-08 14:57:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8655a39cf916044980d94ebe1844697ff6a96a2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-09 09:29:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a6680f207740c7e3df3b9c4f3de3ff080b39b02d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-09 09:30:17.000000000","message":"Patch Set 2: Workflow+1\n\nRebased to unblock the change. Already +W\u0027d by efried","accounts_in_message":[],"_revision_number":2},{"id":"b24b895e26445d65e19cdcbe93655796c99032db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:42:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/76/675276/2/check/openstack-tox-docs/a68e8ff/html/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://logs.opendev.org/76/675276/2/check/openstack-tox-pep8/5dae715/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"68886966c93dc6d5a2e3575314c248751708a1de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:42:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6976f3a46217039a4d39763f8f73f91e0a0d9723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:57:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"322632d2aff2fba045d5734482f046b9de5675ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:57:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/76/675276/2/gate/openstack-tox-docs/4fd0429/html/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://logs.opendev.org/76/675276/2/gate/openstack-tox-pep8/01bb0d2/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb61141e6b2a3cad901535ef090b0729964c1d17","revisions":{"fd9eccdbf56ad1a35d715fedce6d54b0bc89b177":{"kind":"REWORK","_number":1,"created":"2019-08-08 09:13:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/675276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/76/675276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/1"}}},"commit":{"parents":[{"commit":"0fc90beec57d5f0001ef2c85b4c414b4c41b28c9","subject":"Additional upgrade clarifications for cpu-resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fc90beec57d5f0001ef2c85b4c414b4c41b28c9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-08-08 09:13:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-08-08 09:13:02.000000000","tz":60},"subject":"Correct nit with cpu-resources","message":"Correct nit with cpu-resources\n\nChange-Id: Iaff9b643bdfafb9ec26ac430bebb1a094816ef85\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fd9eccdbf56ad1a35d715fedce6d54b0bc89b177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fd9eccdbf56ad1a35d715fedce6d54b0bc89b177"}]},"branch":"refs/heads/master"},"bb61141e6b2a3cad901535ef090b0729964c1d17":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-09 09:29:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/675276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/76/675276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/76/675276/2"}}},"commit":{"parents":[{"commit":"e3d3caef31e0028a89bbdc4766dea78058bef596","subject":"Fix URL to blueprint for Secure Boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3d3caef31e0028a89bbdc4766dea78058bef596"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-08-08 09:13:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-08-09 09:29:57.000000000","tz":0},"subject":"Correct nit with cpu-resources","message":"Correct nit with cpu-resources\n\nChange-Id: Iaff9b643bdfafb9ec26ac430bebb1a094816ef85\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bb61141e6b2a3cad901535ef090b0729964c1d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bb61141e6b2a3cad901535ef090b0729964c1d17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
