)]}'
{"id":"openstack%2Fnova-specs~682511","triplet_id":"openstack%2Fnova-specs~master~I23babc6379fb0602c28337d30b7298bb526802b8","project":"openstack/nova-specs","branch":"master","topic":"bp/add-user-id-field-to-the-migrations-table","hashtags":[],"change_id":"I23babc6379fb0602c28337d30b7298bb526802b8","subject":"Track user_id/project_id for migrations","status":"MERGED","created":"2019-09-17 00:19:21.000000000","updated":"2019-09-19 00:14:19.000000000","submitted":"2019-09-19 00:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682511-1568851966262-805a2d6b","meta_rev_id":"158a0646c19d5acda3d29cd9dbb6a7bdaa09832b","_number":682511,"virtual_id_number":682511,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-09-17 23:57:09.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-09-19 00:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-09-18 23:56:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-09-18 23:56:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 20:26:00.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-18 23:56:55.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-19 00:12:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6280f9d655e663b08c51ae5f099c95112500eec0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-17 00:19:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2e836a933ee8fcbebe6d6d027f50314389f333c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-17 00:20:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1121f189ac5fcdf98e316adb6b74b87d51961ec7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:25:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d4121601934ec8a0675e3d2c15915b : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68274c2928814f30b8ec54271c82537a : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8e5c9ba2614640c62fbb8c8806635bc5fcb1c4e1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-17 00:38:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b3c6ccd0e26f6cdfdff4a842e8646ced125567ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:48:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2631b620d64233992e155a96f86870 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e99b1f1835b8499b9b71c16c837bc3dd : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"03042f56888b4b695102146a8c70246e5e4a0f4d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 20:26:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f827937d1166ef1c197b6a23f2c87a3e56b05b59","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-17 23:56:53.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a76152797c82f1409c3953562f71d23b70448faf","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-17 23:57:09.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThanks Matt.","accounts_in_message":[],"_revision_number":4},{"id":"fb3d7b9251c53ac1c5768b30437d48e87e0c9f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 00:13:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd99213f28b9477a84a206e69649ffe0 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b306733b01c04b0a976e53eed3d04bc1 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":4},{"id":"a50b1574e20f71091be24872c4b050832e7ac6b4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-18 23:56:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nefried@efried-ThinkPad-W520:~/openstack/nova-specs$ diff specs/train/approved/add-user-id-field-to-the-migrations-table.rst specs/ussuri/approved/add-user-id-field-to-the-migrations-table.rst\n 248c248\n \u003c       :header-rows: 1\n ---\n \u003e    :header-rows: 1\n 253a254,255\n \u003e    * - Ussuri\n \u003e      - Re-proposed\n\nFast approving per http://specs.openstack.org/openstack/nova-specs/readme.html#previously-approved-specifications","accounts_in_message":[],"_revision_number":4},{"id":"bed42dbf141990949c29d14a2160f2fca13b656d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 23:57:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3ada17b0bbabfda8942bfbee70ed6ee5f8324a3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:12:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2ff57ee6dcb5c93b069dc3ef14b6afd8c65a0719","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:12:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3560146769243518b366c6b86e6231b : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef94b689a6694a79bc82cf23acb6ea5f : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":4},{"id":"9c2d975d956904859941836af48b555ea2f76c2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:14:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/8b44203316e940ae8f55b3c143866b9f : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"59f1a09f36e19081ad9824a53015ea7b7b1cdac3","revisions":{"9fa52c5e9fec914e895fff0fb79cbb42ac51cc3b":{"kind":"REWORK","_number":1,"created":"2019-09-17 00:19:21.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/11/682511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/682511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/1"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-17 00:11:37.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-17 00:11:41.000000000","tz":480},"subject":"Track user_id/project_id for migrations","message":"Track user_id/project_id for migrations\n\nPreviously-approved: Train\n\nBlueprint: add-user-id-field-to-the-migrations-table\n\nChange-Id: I23babc6379fb0602c28337d30b7298bb526802b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fa52c5e9fec914e895fff0fb79cbb42ac51cc3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fa52c5e9fec914e895fff0fb79cbb42ac51cc3b"}]},"branch":"refs/heads/master"},"cecc93222051a5974dbe99c089b1773b5ad7b5e4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-17 00:20:04.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/11/682511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/682511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/2"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-17 00:11:37.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-09-17 00:19:55.000000000","tz":0},"subject":"Track user_id/project_id for migrations","message":"Track user_id/project_id for migrations\n\nThe blueprint proposes tracking the user_id and project_id of the user that\ninitiated a server migration and exposing those values in the API.\n\nPreviously-approved: Train\n\nBlueprint: add-user-id-field-to-the-migrations-table\n\nChange-Id: I23babc6379fb0602c28337d30b7298bb526802b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cecc93222051a5974dbe99c089b1773b5ad7b5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cecc93222051a5974dbe99c089b1773b5ad7b5e4"}]},"branch":"refs/heads/master"},"1a3ff0653e2865321ef12d89edc360f759ebe948":{"kind":"REWORK","_number":3,"created":"2019-09-17 00:38:14.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/11/682511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/682511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/3"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-17 00:11:37.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-09-17 00:33:43.000000000","tz":0},"subject":"Track user_id/project_id for migrations","message":"Track user_id/project_id for migrations\n\nThe blueprint proposes tracking the user_id and project_id of the user that\ninitiated a server migration and exposing those values in the API.\n\nPreviously-approved: Train\n\nBlueprint: add-user-id-field-to-the-migrations-table\n\nChange-Id: I23babc6379fb0602c28337d30b7298bb526802b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a3ff0653e2865321ef12d89edc360f759ebe948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a3ff0653e2865321ef12d89edc360f759ebe948"}]},"branch":"refs/heads/master"},"59f1a09f36e19081ad9824a53015ea7b7b1cdac3":{"kind":"REWORK","_number":4,"created":"2019-09-17 23:56:53.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/11/682511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/682511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/682511/4"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-17 00:11:37.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-09-17 23:56:32.000000000","tz":0},"subject":"Track user_id/project_id for migrations","message":"Track user_id/project_id for migrations\n\nThe blueprint proposes tracking the user_id and project_id of the user that\ninitiated a server migration and exposing those values in the API.\n\nPreviously-approved: Train\n\nBlueprint: add-user-id-field-to-the-migrations-table\n\nChange-Id: I23babc6379fb0602c28337d30b7298bb526802b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/59f1a09f36e19081ad9824a53015ea7b7b1cdac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/59f1a09f36e19081ad9824a53015ea7b7b1cdac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
