)]}'
{"id":"openstack%2Fnova-specs~682637","triplet_id":"openstack%2Fnova-specs~master~I42d1829ab02db5f927a6bd63235cec667d416264","project":"openstack/nova-specs","branch":"master","topic":"bp/nova-cyborg-interaction","hashtags":[],"change_id":"I42d1829ab02db5f927a6bd63235cec667d416264","subject":"Re-proposed Nova Cyborg interaction specification.","status":"MERGED","created":"2019-09-17 13:01:32.000000000","updated":"2019-09-27 22:10:11.000000000","submitted":"2019-09-27 22:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682637-1569622124186-953ecec0","meta_rev_id":"8efdb9fa3083d3e91650fc2a3bf1e8077f36f227","_number":682637,"virtual_id_number":682637,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 22:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-09-27 21:43:31.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-09-27 21:43:31.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 15:18:36.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-09-25 15:18:44.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-25 15:18:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-09-25 15:19:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-09-25 15:19:15.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-09-27 21:43:31.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-27 22:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79c2984181717f53375927a54507e4e502cf65bd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-17 13:01:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d8c6a3dfcaa861a6fa33cdebb924b5c2641ebb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:13:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/786d5528227d48d1a4f7096c2ff69f36 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c557300c174186969c2639569e5702 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"82b5d1a9b68e21f38a712e5411b36ad65de3f279","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-23 21:42:24.000000000","message":"Topic set to bp/nova-cyborg-interaction","accounts_in_message":[],"_revision_number":1},{"id":"8806010d00a32a9907813877902b3c67f8e45b54","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-23 21:51:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nThe delta from the Train approved spec is nontrivial. Examining the changes closely, they seem to be fairly simple updates to reflect a couple of issues that were resolved during the last cycle, and to remove a bunch of references that are no longer pertinent.\n\nFor future reference, it makes it easier for reviewers in such cases if you upload the first patch set identical to the previously approved spec, and then a second patch set with the changes.\n\nIn this case, you may wish to scale back this change set to *only* s/Train/Ussuri/ and enhance the History table, and then propose a separate change set for the deltas. The former would be suitable for fast approval per [1], and the latter should be fairly simple to get approved subsequently.\n\n[1] http://specs.openstack.org/openstack/nova-specs/readme.html#previously-approved-specifications","accounts_in_message":[],"_revision_number":1},{"id":"206edc56af1aba8c5fef5f80115e3cc23430c1e4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-24 00:03:23.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThanks, Eric.\n\n\n \u003e In this case, you may wish to scale back this change set to *only*\n \u003e s/Train/Ussuri/ and enhance the History table, and then propose a\n \u003e separate change set for the deltas. \n\nDone.","accounts_in_message":[],"_revision_number":1},{"id":"284f833149fb76e173581c0578eb9ee5851f95e9","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-24 00:28:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f410087a98c4ef5842e39d73b9d13936c3c84f68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 00:44:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0dbc4620ec94f8c89038ff6ca67fc04 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da85d3cb3105472c8b822d2f588ceb52 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5375a7b832e246a21c1120d13e6605ecd41e2463","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-27 21:43:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a1cf54a23f86a6e5e6cbf647e0749c3e0debfe64","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-27 21:43:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nFast approving per http://specs.openstack.org/openstack/nova-specs/readme.html#previously-approved-specifications","accounts_in_message":[],"_revision_number":3},{"id":"b7c6003835cccf85b3c6f579cded858cb481b0a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 21:57:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec814d2d99c4f6194d5d7b77fa37e83 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354c5bb7863c40db9f3a576304d3ce55 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":3},{"id":"68fbfc3a19f52ccf35580699d3f4afe654375fe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 21:57:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d19c5108ef6f372c041ac00f80e88f67c88f3857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 22:08:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecfb28e65f104b8f84a0a5e16c6e1e1f : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8aa994143f4fe19099917de3c32465 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"a42afbbbad0a8f97b8760204e558263743758fa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 22:08:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8efdb9fa3083d3e91650fc2a3bf1e8077f36f227","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 22:10:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/48bdef6b6b774718b92eab1ab9e8c49b : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a11b0d5b4f33c277d97bf0c70aa789956b712c78","revisions":{"a89dc0c168fedc02453182ffa3f1d42cc9b4c0f2":{"kind":"REWORK","_number":1,"created":"2019-09-17 13:01:32.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/37/682637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/682637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/1"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 14:58:26.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 14:58:26.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: I42d1829ab02db5f927a6bd63235cec667d416264\nBlueprint: https://blueprints.launchpad.net/nova/+spec/nova-cyborg-interaction\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a89dc0c168fedc02453182ffa3f1d42cc9b4c0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a89dc0c168fedc02453182ffa3f1d42cc9b4c0f2"}]},"branch":"refs/heads/master"},"f0862c095168b354ef7f9e2a9e94cc73844258b2":{"kind":"REWORK","_number":2,"created":"2019-09-24 00:28:38.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/37/682637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/682637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/2"}}},"commit":{"parents":[{"commit":"c0155ebb1c28ddeaea43863f45de0576783ef2e3","subject":"Merge \"resubmit image metadata prefiltering spec for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c0155ebb1c28ddeaea43863f45de0576783ef2e3"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 14:58:26.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-23 23:55:57.000000000","tz":-420},"subject":"Re-proposed Nova Cyborg interaction specification.","message":"Re-proposed Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nNo changes from Train version except History.\n\nChange-Id: I42d1829ab02db5f927a6bd63235cec667d416264\nBlueprint: nova-cyborg-interaction\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0862c095168b354ef7f9e2a9e94cc73844258b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f0862c095168b354ef7f9e2a9e94cc73844258b2"}]},"branch":"refs/heads/master"},"a11b0d5b4f33c277d97bf0c70aa789956b712c78":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-27 21:43:13.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/37/682637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/37/682637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/37/682637/3"}}},"commit":{"parents":[{"commit":"c0155ebb1c28ddeaea43863f45de0576783ef2e3","subject":"Merge \"resubmit image metadata prefiltering spec for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c0155ebb1c28ddeaea43863f45de0576783ef2e3"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-17 14:58:26.000000000","tz":-420},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-27 21:43:10.000000000","tz":0},"subject":"Re-proposed Nova Cyborg interaction specification.","message":"Re-proposed Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nNo changes from Train version except History.\n\nPreviously-approved: Train\nChange-Id: I42d1829ab02db5f927a6bd63235cec667d416264\nBlueprint: nova-cyborg-interaction\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a11b0d5b4f33c277d97bf0c70aa789956b712c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a11b0d5b4f33c277d97bf0c70aa789956b712c78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
