)]}'
{"id":"openstack%2Fnova-specs~682655","triplet_id":"openstack%2Fnova-specs~master~I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57","project":"openstack/nova-specs","branch":"master","topic":"bp/flavor-extra-spec-validators","hashtags":[],"change_id":"I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57","subject":"Re-propose the flavor extra spec validation spec","status":"MERGED","created":"2019-09-17 14:04:22.000000000","updated":"2020-01-30 15:51:29.000000000","submitted":"2020-01-30 15:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":105,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682655-1580399177195-3783239c","meta_rev_id":"ea2bbdcdf6a7ea54e97ee08856f6e9001c1978f6","_number":682655,"virtual_id_number":682655,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2020-01-30 13:02:58.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 15:46:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2020-01-30 12:54:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-01-30 15:33:12.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-30 13:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2020-01-30 15:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-01-30 15:33:12.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 00:01:48.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-01-30 12:54:51.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-01-30 13:12:20.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-30 15:33:12.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-30 15:38:49.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-01-30 15:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fc96c5d802365abdaef04d037c5c412152ad8f3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-17 14:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a9142cf42098d8923d570ecc6ec5e561e801d33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:17:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e52d0c2c6644c0ba38c23c6aad3e77 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7afea8b1a44730a9be92fdfab207a8 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cacbe21768008545fcc56b1eda88a3922c7938e2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-19 00:06:44.000000000","message":"Patch Set 1:\n\nefried@efried-ThinkPad-W520:~/openstack/nova-specs$ diff specs/train/approved/flavor-extra-spec-image-property-validation-extended.rst specs/ussuri/approved/flavor-extra-spec-validators.rst \n 7,9c7,9\n \u003c \u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n \u003c Flavor Extra Spec Validation (Extended)\n \u003c \u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n ---\n \u003e \u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n \u003e Flavor Extra Spec Validator\n \u003e \u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n 11c11\n \u003c https://blueprints.launchpad.net/nova/+spec/flavor-extra-spec-image-property-validation-extended\n ---\n \u003e https://blueprints.launchpad.net/nova/+spec/flavor-extra-spec-validators\n 13,14c13\n \u003c Introduce a YAML-based schema describing flavor extra specs, image metadata\n \u003c properties, and the relationship between each.\n ---\n \u003e Introduce a pluggable system to describe and validate flavor extra specs.\n 22,27c21,23\n \u003c - Lack of documentation for many flavor extra specs and image metadata\n \u003c   properties [1]_. Yes, we have Glance image metadata definitions but they\u0027re\n \u003c   generally rather out-of-date, we don\u0027t/can\u0027t consume them in Nova, and\n \u003c   they\u0027re not aimed towards user-facing documentation.\n \u003c \n \u003c - Outdated, incomplete or incorrect Glance metadata definitions.\n ---\n \u003e - Lack of documentation for many flavor extra specs [1]_. While, Glance has\n \u003e   metadefs [2]_, those are generally out-of-date, incomplete, and not\n \u003e   consumable from nova and our user-facing documentation.\n 78c74\n \u003c \u003chttp://specs.openstack.org/openstack/nova-specs/specs/stein/approved/flavor-extra-spec-image-property-validation.html\u003e`,\n ---\n \u003e \u003chttp://specs.openstack.org/openstack/nova-specs/specs/stein/implemented/flavor-extra-spec-image-property-validation.html\u003e`__,\n 358a355,357\n \u003e 5. Add a tool to generate glance-metadef compatible JSON files that can be\n \u003e    consumed by the glance metadata definitions catalog API.\n \u003e \n 381a381\n \u003e .. [2] https://github.com/openstack/glance/blob/18.0.0/etc/metadefs/compute-libvirt.json\n 393a394,395\n \u003e    * - Ussuri\n \u003e      - Re-proposed with a simpler name and minor modifications\n\nPlease create (or rename existing) blueprint for paperwork purposes, and I\u0027ll have a closer look at this delta later-ish.","accounts_in_message":[],"_revision_number":1},{"id":"ee5c248be1dc417c5e173a0b8a569a8304e6d900","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-10-23 00:01:48.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nThanks for the detailed spec, Stephen.","accounts_in_message":[],"_revision_number":1},{"id":"fd1f75db8ff48f4cfc0748e4e2e2c686489b54c8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-27 13:45:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nFor me this spec is OK but we have interesting questions from Sundar so I\u0027m holding my +2","accounts_in_message":[],"_revision_number":1},{"id":"0a2aa3d34ab27ff63eb5d8d69783701c604fc78f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-27 15:23:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"35d0cc5576b4b25f3051beefacd34707d351bf75","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 17:16:55.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"437597a0732c55498b18c7f029baad80cafd0ee1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-28 00:27:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(16 comments)\n\nI think this is a noble effort, but IMO the philosophy around external definitions needs to be crisped up, and I\u0027m -1 on the whole stevedore concept unless we have an immediate need for validation more powerful than regex. Other issues, some nontrivial, inline.","accounts_in_message":[],"_revision_number":1},{"id":"f3faf503e3aa5d5a2abc20898b6ac266d1cc8b83","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-28 09:05:48.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"faa8c569ae460dcdc3294627cae39dbb9c59ecd6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-28 14:47:41.000000000","message":"Patch Set 1: Workflow-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f1fbfe8ca85c2c6b23ba0ca71d38f5dc07656616","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-28 14:54:48.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"155d93f19004e722333850ee8ac41d8ed9b7d844","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-28 15:00:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d106b4d13c9bef31d1a2caba0e162f4f30a5edb6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-28 15:09:12.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nSorry, we crossed in the mail on that last one, further responses below.","accounts_in_message":[],"_revision_number":1},{"id":"e35ad448d03c684fd9cc363069c41d0543e81fca","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-29 09:31:06.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"f008b7031b7e794d5f03060392f065554113aab5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 10:36:11.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"985d01a947d744bbd734fed505aee81938338fcc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 11:08:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24be563dc737d8aae8cd7a41bf6be91e504e4f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:18:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fd594e4bfb41eca2cdad7431639f72 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda996f4ac474a6c8cb0954103d5d896 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"fdad7aa0455c631b3d2bec0730016a810f1d3080","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 11:40:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"35d01c71903182019a5731413e69b9a80ccf3a64","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 11:43:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"348de75c454a5a6ab0bb0a7ba1d7ac29fe389fca","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-29 11:47:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"435a819f8f6af6ea01444b20a9471de1c5e6834e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:55:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112b441e88ec4db68fd9c657cc30ba07 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f195e811d9d54c8489879499e94c5d3f : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c5b591e169eac6676d76ee5fa9efe9d2c12b7496","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 12:19:19.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1f3082395c15e547732a3ac1969773cc9b815279","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 12:32:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0a12be26b687aeacbab0fb96a8226280e6b06491","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 12:35:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d404371b00c993ceaa7301f0bd20205c3d058a3c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 13:05:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2906311a27955580b10f3bf8407a686abcd294e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 13:14:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d3b16cf3cefafb1cea35d01682fa2330c1cc699a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-29 14:17:24.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82abda2177113bfcc507c597b357e1e3d6df94b4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 14:44:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e438d22a8fe7a92326e07aadf6955f6d5cd6bccc","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-29 14:57:41.000000000","message":"Patch Set 3:\n\n(5 comments)\n\nI haven\u0027t actually read PS3 yet, just continuing conversations in comments so far.","accounts_in_message":[],"_revision_number":3},{"id":"5a096ec30ce81d47111f12ec32962ec8d4e24312","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-29 15:20:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"430d4059560b7aeb94e503db2caf54887d6f9826","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 17:01:59.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f13282c1c363ca4168159e5206325aa32e377f61","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 21:36:08.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nkeeping -1 until this is updated with the new query argument for contoling validation but you where right the cpus is a mask not a count.","accounts_in_message":[],"_revision_number":3},{"id":"993795147994088463fafe4d279061f3f8a167df","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-29 21:50:04.000000000","message":"Patch Set 3: Code-Review-1\n\nLeaving a vote to note that what we agreed upon today still needs to be folded in.\n\nDiscussed in IRC [1] and agreed upon the following points:\n- No YAML. In-tree stuff in python; external stuff by stevedore only.\n- In the new microversion, include a queryparam that lets us twiddle the level of validation among (names tbd) a) strict (the default), which bounces on bad values *and* unrecognized keys; b) permissive, which bounces on bad values but ignores unrecognized keys; c) off, which doesn\u0027t enforce any rules, but (maybe) logs messages for violations.\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2020-01-29.log.html#t2020-01-29T14:58:26","accounts_in_message":[],"_revision_number":3},{"id":"104bbc360f61a53618e53aa60d9509c2ed0a43c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:09:29.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThink the new version accurately reflects our discussions on IRC yesterday","accounts_in_message":[],"_revision_number":3},{"id":"5a17ff68975136274f803b3f70cad8e243f4e12f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:09:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2dc7529afe94a421ab3e83b98229e3c79e80b38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:21:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011d0c38aaed4c39a3658177e91fb4db : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7b88e0e5fe4075af1cd02cbfb5bb62 : FAILURE in 5m 35s","accounts_in_message":[],"_revision_number":4},{"id":"9790c175819382d64e81ef2ab97cc439151b32ca","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:43:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c8ac45139b984bc513f1bce196b2f07d6426f929","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-30 12:54:51.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)\n\nassuming this passes pep8 then ya this looks good to me","accounts_in_message":[],"_revision_number":5},{"id":"50aa753b34f00fd6788ec5d821b4002c5f386388","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 12:58:20.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nI have one new questions. But overall it is OK to me.\n\ngibi","accounts_in_message":[],"_revision_number":5},{"id":"35a066cf00992e12ccfe11484312643e3f1ddcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:00:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9003bd16149343288f21413ea5ccf104 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c080b9d3c8b4e24b2e358bfda8359ac : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":5},{"id":"9162353041383fe4af6daadf2fefc9bf95592714","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 13:02:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cc1b75e62d0b426c5b6f7114a036721ceb7e570b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 13:12:20.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7998a28e59cccf0b4146c480762c770bbd21e667","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-30 15:33:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(10 comments)\n\nThere are a few minor clarifications that could be made in a fup, plus hints at some further discussion to be had on the implementation, but this captures what we agreed on yesterday. Let\u0027s do it.","accounts_in_message":[],"_revision_number":5},{"id":"39af4ba8b2d4bb5b6a7524f6483cecada86d42b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:33:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ca5035aeed1f89f474727c1b5c067e647eb673b2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-30 15:38:49.000000000","message":"Patch Set 5: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fa75d5acae7cc497ebb8f9b244cf033c22f21650","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:46:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4dd36c1a88232ed87f837cd125c2c53b7c590dbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:46:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b1c86c606547928bec1dc0a05209ff : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54cf03f2edf943c883e3fc90ef638508 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":5},{"id":"58b8257133a6eb787580463233986d4458060b79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:51:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e4afb17939664fab88998a65ea00e210 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"90f006e96c2c42d10e1ce8b78d514ec9f15b670b","revisions":{"45e145a1da9e2ae445041d1018dd5d03c144b797":{"kind":"REWORK","_number":1,"created":"2019-09-17 14:04:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/682655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/682655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/1"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-17 13:54:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-17 13:59:20.000000000","tz":60},"subject":"Re-propose the flavor extra spec validation spec","message":"Re-propose the flavor extra spec validation spec\n\nPreviously known as the (rather lengthy)\n\u0027flavor-extra-spec-image-property-validation-extended\u0027 spec.\n\nPart of bp flavor-extra-spec-validators\n\nChange-Id: I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45e145a1da9e2ae445041d1018dd5d03c144b797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45e145a1da9e2ae445041d1018dd5d03c144b797"}]},"branch":"refs/heads/master"},"d0b3c4d3b86a6869d802d4ccb50b5c22c89c0a21":{"kind":"REWORK","_number":2,"created":"2020-01-29 11:08:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/682655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/682655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/2"}}},"commit":{"parents":[{"commit":"481a6ee726d68a03a04eb505e5affe2bab8c59d2","subject":"Merge \"support live migration with virtual persistent memory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/481a6ee726d68a03a04eb505e5affe2bab8c59d2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-17 13:54:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-29 11:07:52.000000000","tz":0},"subject":"Re-propose the flavor extra spec validation spec","message":"Re-propose the flavor extra spec validation spec\n\nPreviously known as the (rather lengthy)\n\u0027flavor-extra-spec-image-property-validation-extended\u0027 spec.\n\nPart of bp flavor-extra-spec-validators\n\nChange-Id: I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0b3c4d3b86a6869d802d4ccb50b5c22c89c0a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0b3c4d3b86a6869d802d4ccb50b5c22c89c0a21"}]},"branch":"refs/heads/master"},"afb4ad2ae4068c8af0467a93e627c6434b7f767a":{"kind":"REWORK","_number":3,"created":"2020-01-29 11:43:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/682655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/682655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/3"}}},"commit":{"parents":[{"commit":"481a6ee726d68a03a04eb505e5affe2bab8c59d2","subject":"Merge \"support live migration with virtual persistent memory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/481a6ee726d68a03a04eb505e5affe2bab8c59d2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-17 13:54:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-29 11:43:28.000000000","tz":0},"subject":"Re-propose the flavor extra spec validation spec","message":"Re-propose the flavor extra spec validation spec\n\nPreviously known as the (rather lengthy)\n\u0027flavor-extra-spec-image-property-validation-extended\u0027 spec.\n\nPart of bp flavor-extra-spec-validators\n\nChange-Id: I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/afb4ad2ae4068c8af0467a93e627c6434b7f767a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/afb4ad2ae4068c8af0467a93e627c6434b7f767a"}]},"branch":"refs/heads/master"},"96b4186a0f801c287ab77ca2390c60c3b9f426a1":{"kind":"REWORK","_number":4,"created":"2020-01-30 12:09:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/682655/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/682655/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/4"}}},"commit":{"parents":[{"commit":"481a6ee726d68a03a04eb505e5affe2bab8c59d2","subject":"Merge \"support live migration with virtual persistent memory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/481a6ee726d68a03a04eb505e5affe2bab8c59d2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-17 13:54:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-30 12:09:37.000000000","tz":0},"subject":"Re-propose the flavor extra spec validation spec","message":"Re-propose the flavor extra spec validation spec\n\nPreviously known as the (rather lengthy)\n\u0027flavor-extra-spec-image-property-validation-extended\u0027 spec.\n\nPart of bp flavor-extra-spec-validators\n\nChange-Id: I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/96b4186a0f801c287ab77ca2390c60c3b9f426a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/96b4186a0f801c287ab77ca2390c60c3b9f426a1"}]},"branch":"refs/heads/master"},"90f006e96c2c42d10e1ce8b78d514ec9f15b670b":{"kind":"REWORK","_number":5,"created":"2020-01-30 12:43:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/682655/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/682655/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/682655/5"}}},"commit":{"parents":[{"commit":"481a6ee726d68a03a04eb505e5affe2bab8c59d2","subject":"Merge \"support live migration with virtual persistent memory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/481a6ee726d68a03a04eb505e5affe2bab8c59d2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-17 13:54:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-30 12:29:02.000000000","tz":0},"subject":"Re-propose the flavor extra spec validation spec","message":"Re-propose the flavor extra spec validation spec\n\nPreviously known as the (rather lengthy)\n\u0027flavor-extra-spec-image-property-validation-extended\u0027 spec.\n\nPart of bp flavor-extra-spec-validators\n\nChange-Id: I1b3b74f1fd2af9419f7aa0c79b5dad5b26065b57\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/90f006e96c2c42d10e1ce8b78d514ec9f15b670b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/90f006e96c2c42d10e1ce8b78d514ec9f15b670b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
