)]}'
{"id":"openstack%2Fnova-specs~685857","triplet_id":"openstack%2Fnova-specs~master~Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6","project":"openstack/nova-specs","branch":"master","topic":"bp/whose","hashtags":[],"change_id":"Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6","subject":"Add \u0027Feature Liaison\u0027 spec process","status":"MERGED","created":"2019-09-30 22:59:12.000000000","updated":"2019-10-17 00:04:56.000000000","submitted":"2019-10-16 14:43:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":75,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685857-1571236980750-f9e06b29","meta_rev_id":"98b3b96d92480ea7363d5a8b745df1f3b9961ec8","_number":685857,"virtual_id_number":685857,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 14:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-14 16:55:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-10-15 07:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-10-16 14:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-10-16 14:32:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 13:44:48.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-10-04 15:28:39.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2019-10-08 07:26:07.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-10-11 07:52:55.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-10-14 16:55:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-10-15 07:25:14.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-10-16 14:32:06.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-10-16 14:43:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55fe20fa32818ea93c5a8d045092a1203070abfd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-30 22:59:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3df55b4b971b258892db87f9b53c35a38c4daa9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 23:12:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0a7ca681c34b288c7311f3b492ad2a : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6596d8f05584d6c8211f9cd48c0b6a6 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a0d44b60b83f7795f46d17d5f11e32634ac09585","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-10-01 11:17:57.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI understand the problem (we all do), and I like this idea as a solution that we could try.\n\nMy comments inline are about not needing an actual core for some of the tasks.\n\nCulture stuff like \"no one\u0027s going to look at your patch if it\u0027s WIP, missing unit tests, or is failing CI (for reasons real or fictional)\" could be written down somewhere - perhaps even linked from that \"welcome first-time contributor\" message - I dunno if something per-project like that could be possible in Gerrit.\n\nThe more nuanced stuff could still be done by a non-core as an initial pass. There are certain community members that, while not cores and not being able to +2 a thing, could probably do a decent enough job of *refusing* a thing because it\u0027s obviously not mergeable. We would need to identify who these people are, make sure they\u0027re actually willing to \"officially\" take on this responsibility, and keep a list of these \"initial liasion\" people somewhere.\n\nThough, pushing this idea further - do such people actually exist? They\u0027re most likely either cores or on their way to becoming cores, so maybe this is all pointless.","accounts_in_message":[],"_revision_number":1},{"id":"8a2bb84eaa9dab8afe55c4cdb49cbf04a51ee024","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-01 14:24:06.000000000","message":"Patch Set 1:\n\nThanks Artom.\n\n\u003e could be written down somewhere\n\nAgree, let\u0027s do that, BUT nobody reads that stuff. Linking from the first-time contributor thing would help a little, but nova culture has its own nuances (and let\u0027s be frank, some of those nuances I wouldn\u0027t want written down). What this does is put it right in the dev path so you can\u0027t miss it.\n\n\u003e not needing an actual core\n\nAgree again; the problem is finding a way to phrase that and identify who qualifies and and and. I wanted to KISS for the sake of the template, and we can allow non-core liaisons on a case by case. I\u0027ll try to add some words allowing for that.","accounts_in_message":[],"_revision_number":1},{"id":"af565275485230a59aeadac130489de5488de833","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-01 14:34:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2f0ed26bd5ea7ab3fb07ef688cd35890869f43f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:51:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a323f106bc3408ea82644abe0d24048 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bbe23e3c444ae59c867963877f8dce : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"7b1a720aac799b527dc4158e1e70ca7ee839f71e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-10-01 15:16:21.000000000","message":"Patch Set 2: Code-Review+1\n\nI\u0027m on board with this - and I agree that we should avoid bikeshedding all the intentional vagueness out of this, try it for a cycle or two, and see how it goes.","accounts_in_message":[],"_revision_number":2},{"id":"9a6bcbba547eee808711ab7505ab81e23ec4eba7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-01 22:11:52.000000000","message":"Patch Set 2: Code-Review+1\n\nnice idea.","accounts_in_message":[],"_revision_number":2},{"id":"b4b8102cafe9d140f19054145d8637b615e4a1b1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-02 18:14:50.000000000","message":"Patch Set 2:\n\nTwo things:\n\n1. We do have quite a bit of contributor documentation that covers the things Artom brought up:\n\nhttps://docs.openstack.org/nova/latest/contributor/process.html\n\nBut I agree with Eric that people mostly likely aren\u0027t going to read all of that. I think if someone were so inclined, they\u0027d have a recorded youtube video talking with some cores and contributors about what it takes to land code in nova, in the format like we used to have years ago when sdague ran the bootstrapping hour videos:\n\nhttps://www.youtube.com/channel/UCaw2N0DcGtONDMGz_PGtBJw\n\nThen you just say for new contributors docs, here, go watch this 30 minute video. Having said that, I\u0027m sure there are summit session videos already about contributing to openstack and I don\u0027t know if people read them. When I was new to openstack I had mentors already involved in nova that helped me along (cyeod, sdague, dansmith) so it was much easier but those were also wildly different times where there were a lot of diverse vendors who had enough full time upstream people that could spend time doing that mentoring.\n\n2. On the subject of knowing who to ping, we also have some older things like:\n\nhttps://wiki.openstack.org/wiki/Nova#People\n\nhttps://wiki.openstack.org/wiki/Nova/BugTriage#Tag_Owner_List\n\nThose are likely wildly out of date nor am I sure about their effectiveness.","accounts_in_message":[],"_revision_number":2},{"id":"fb08ade890c2ab0140d16006a3506d2451369e76","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-02 21:40:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45f2fa51154c9fc3ca673125133251be25e64d9d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-02 23:47:49.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nGoing to take out the \"optional\" bit.\n\n@Matt: yes to all of that, but mainly the part where all those resources are some combination of gone, unknown, unused, unmaintained, and hopelessly outdated. Putting this into the spec process gets it in front of every person proposing a spec, every time, so it\u0027s at least as fresh as whenever the spec is authored.","accounts_in_message":[],"_revision_number":2},{"id":"2a6292ba92b9b5c3b310803edfd6b0b72845e9ac","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 15:09:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d9de8c9bc02055d0c3176c20b047080ec3d56a17","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 15:10:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ae47ce3ba1a27424d899db808856a90b9784641","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 15:11:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b9e154f2aa9f18b9ffbd2bad5dabda4de6f90998","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-03 15:20:08.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bf952f1c6eba5f48eb1f39dd50c91f27a02a5331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:24:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee77586edfb841bb82f8c052d2583749 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ceccdcd0ce4f0c8be43c062f34c7a9 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"be65fca1c77c45363f7b795b12878b51d41ba1e6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 15:41:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b5d56fcb129b1c32f7ef59e75c3a0155df7343f6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-03 15:52:03.000000000","message":"Patch Set 3: Code-Review-1\n\nGiven the confusion over what a sponsor is (designated hand-holder to newbs, on the hook reviewer, both?) I definitely think we need something documented in the review process docs with this change:\n\nhttps://specs.openstack.org/openstack/nova-specs/readme.html#specification-review-policies\n\nSince it\u0027s confusing at least a couple of us (Dan and myself) about what we\u0027d be signing up for if we sponsor something, and what it means for a core that approves a spec but doesn\u0027t officially sponsor it, i.e. you\u0027re probably a jerk.","accounts_in_message":[],"_revision_number":3},{"id":"888e407c06e0467161cf8e07961d23d17eac63a5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 16:14:15.000000000","message":"Patch Set 3:\n\n\u003e something documented in the review process docs\n\nAck, good call.","accounts_in_message":[],"_revision_number":3},{"id":"e9eabbfa08e5a7bd571c0cd414a39754d8186e0f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-03 16:36:50.000000000","message":"Patch Set 3:\n\nAs sponsor is going towards more for helping the new contributors instead of a dedicated nova-core to merge the code. we should also mention it clearly in doc about it. The sponsor is \"help to get the review done\" and not officially \"responsible for merging the code\". It can be any other member (core or nova experienced developer) which does not neccessary had +2/+A the spec.","accounts_in_message":[],"_revision_number":3},{"id":"16b411370eea33c64b176224be1e7ed511da4a73","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 16:45:13.000000000","message":"Patch Set 3:\n\n\u003e The sponsor is \"help to get the review done\" and not officially \"responsible for merging the code\". It can be any other member (core or nova experienced developer)\n\nAgree, Artom pointed this out as well [1], but I\u0027ll make sure I include it in the doc update.\n\n[1] https://review.opendev.org/#/c/685857/1..2/specs/ussuri-template.rst","accounts_in_message":[],"_revision_number":3},{"id":"0129d3f3de7f7b016bc75e453f73a02a60a5e539","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 21:19:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"838019073d33c8031d8fdbe0a7f18c8e47efbefc","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 21:20:15.000000000","message":"Patch Set 4:\n\nNote to reviewers: This (still, intentionally) doesn\u0027t make any attempt to define how we\u0027re going to cap/throttle work in a cycle.","accounts_in_message":[],"_revision_number":4},{"id":"df96e4997ce0c522a2b91a9dc2993556a70b01e8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 21:21:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ad9f8ddc0e8a39d512161bcd124cba6422f2219d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-03 21:23:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"594158c73ea23a1d21e52e153abbfced357bd585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 21:35:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca9e87d5cd14d31aca9036dd81f305b : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce3debd872646768fc5b85508cb4ec4 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"377039bb9c19b88d673124429dea5bc32c6ea82d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-10-04 07:51:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bc4e48ecd712563ce88f3e648357cddbc7272c51","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-10-04 09:17:13.000000000","message":"Patch Set 4: Code-Review+1\n\n(7 comments)\n\nLooks good to me. Just some nits inline.","accounts_in_message":[],"_revision_number":4},{"id":"2db69d6b960c81c96996d183e92b17cb5194a865","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-04 13:44:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"40dd0b06eacbefb881d6621d0061c842f7f0ec7e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-04 13:58:49.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e3332e742363870834792c3378e88616c82066de","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-04 14:32:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8a7f92f5b355d52736420275e6efb4fc66fe2da7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-04 14:35:16.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"20968bed0f46f9755a7a4d602426cc5b32f3d44e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:51:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8aaf28cd2040359cd4f3d6b055e097 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45726dff9b1d4d57bf62a09f977dbd14 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":5},{"id":"8ea117569609f30ea74a7b47d17063a37cb87c6c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-10-04 15:17:00.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nI\u0027m happy with this.","accounts_in_message":[],"_revision_number":5},{"id":"7825f881e143bd6ba27ac592baeca3a20660669b","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-10-04 15:28:39.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)\n\nBesides the horrendous lack of consistency in people\u0027s nicks, this is good :)","accounts_in_message":[],"_revision_number":5},{"id":"e7fa3c15743a9aef5091ca18036a6d92385b6d33","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-10-04 15:30:04.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0f809b6e2cdb8dc1f81d70a738f1226e0da9e2a5","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-10-04 15:46:21.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nAnother grip inline about overloading the \"core\" term - fix or ignore at your discretion :)","accounts_in_message":[],"_revision_number":5},{"id":"4105bbfe2ead02becb0d2fb97b4db313b2be9889","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-04 16:09:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2b03f26da20f894e030c25f225333ad4fdf4725c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-10-04 16:15:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6aca2bce4969f7470384bba5d88c6261b7936379","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-04 16:42:37.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)\n\nlgtm, thanks","accounts_in_message":[],"_revision_number":5},{"id":"ff31172f2efcddc2d65d97d4e04343435b7aa26e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-10-07 08:01:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nHonestly, I think we could take a bit of time to modify the README to explain more about the fact that the liaison would be rather a SME (Subject Matter expert) than a \u0027core\u0027.\n\nHere, what we want ? To help feature owners to know \"who to could help\" them. That\u0027s it, no need to be a core for helping others :-)","accounts_in_message":[],"_revision_number":5},{"id":"c51d692eb1a2fc4971b03f968c0ba1a82bfea23e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-07 19:35:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7cf39b2f54d71c7ab31b54529911143d13a1a1e0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-07 19:40:43.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1e11fb6489b5fa1ab6b07326fb365aebd70b2f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:49:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ed855a89c34dbd92c392706656c0ca : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02e095fa8fc40b1be067d084e15b1cb : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":6},{"id":"cb48833e2b9175fa6d46054893aaebb14429d0b6","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-10-08 07:26:07.000000000","message":"Patch Set 6:\n\n(2 comments)\n\n+1 for the idea. Hope that is helpful for the people. just asking some question","accounts_in_message":[],"_revision_number":6},{"id":"3a98ba051dd04537597377463bee2f4c45bf3f22","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-10-08 08:38:55.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nI\u0027m basically OK with the feature liaison implementation, but I have serious concerns about some rule enforcements that we\u0027d make upstream and which I think are subject to debate.","accounts_in_message":[],"_revision_number":6},{"id":"2c44859389348fc9320accfd442a90f1971e3aeb","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 15:10:03.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bfd8529f721dc63eff4cd9b9027ab999b32e5ffb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-11 06:47:08.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2bee22cb1e62b7c7b9653d16c5f77731d8dfbd55","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-10-11 07:42:53.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"49f9c5636bc6cf9840010952d2a7b06f4b74bd3e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-11 07:52:55.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0d78d4eba9559cb0a626f19d4cb58082a95493d6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-14 16:25:14.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"481de95a7b50f25e547486a73d32fae7659803c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-14 16:25:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cd29ac59c39e8286250c26e26a9abe9c6c32e226","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-10-14 16:36:41.000000000","message":"Patch Set 7: Code-Review+2\n\nOK, after commenting a lot about this spec process, I think I\u0027m quite +2 with it, even if I could still have some concerns.\nGiven it\u0027s important to have a consensus about this, I think it\u0027s more important to just accept it and then providing other follow-ups if there would be some points.","accounts_in_message":[],"_revision_number":7},{"id":"2be29d556c70047163fa6566146ed78b876f0ac6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-14 16:39:43.000000000","message":"Patch Set 7: Code-Review+1\n\nlgtm, thanks","accounts_in_message":[],"_revision_number":7},{"id":"a06006c8f54f3952c012891756e9cee0d11eb58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 16:40:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62a21fab8c24d90b9d5841f0633cffc : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6baf125a694c6594e96b8752e92444 : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":7},{"id":"668dab8616acea8b45a924275e88bdaebd35cb5e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-14 16:52:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a213af34fc9b89bf64386457e0de292ab1b90c2b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-14 16:55:05.000000000","message":"Patch Set 8: Code-Review+1\n\nlgtm.","accounts_in_message":[],"_revision_number":8},{"id":"305cb413fc8ffbcb1548a12ce060abcabb8d68d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:07:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ee0df07b684d92851e2c6af02ced69 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a2c97f8a3c40ce96d98ce3f21c9618 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":8},{"id":"5b7be09d9cb20535e3b17ebea434da60ddcc130d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-10-15 07:25:14.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"23821b207bb423ff2c658fbe65aac456dbd816ee","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-10-16 14:32:06.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2edc0105ee55fbff07878948a0b0795f14a72454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:32:18.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cc0d8b6a87fe90c5c3d0751f228866dc9c2fbff5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:43:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"1b89daf49131ce4de1bb284f604ff56ee0ad30ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:43:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b6284b51544184b70b1a53ca71d367 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a306345fe3483d94da03da9ca1328e : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":8},{"id":"66b1a8ef95e9f4d3dbb05a651b97319793a74f78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:45:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/fda97127d0474cfcb7e48767e08d2f17 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":8},{"id":"0ccdc0dc21e0e5d4873480fbfdc9dc6bcf4d4c7a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-17 00:04:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a1421755787a6541a322f7add36f987a2920e882","revisions":{"453addd12f6231288a8696e3590390d0f39f7fe2":{"kind":"REWORK","_number":1,"created":"2019-09-30 22:59:12.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/1"}}},"commit":{"parents":[{"commit":"5f16a02137f8fadd4891ff539ff100fa9f358add","subject":"Fix invalid link index","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f16a02137f8fadd4891ff539ff100fa9f358add"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"subject":"Add \u0027Core Liaison\u0027 section to spec template","message":"Add \u0027Core Liaison\u0027 section to spec template\n\nRequire a spec to be sponsored by a member of the core team.\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- \"Non-core contributors don\u0027t understand how nova dev culture works\"-itis\n- \"Nobody but me cares about my feature\" syndrome\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them\" disorder\n\nIn the current commit, this section is optional in ussuri, but I would\nprefer to amend the existing approved specs to add it, and make it\nrequired immediately.\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/453addd12f6231288a8696e3590390d0f39f7fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/453addd12f6231288a8696e3590390d0f39f7fe2"}]},"branch":"refs/heads/master"},"4ef48a18450ab5fc17aa595008457c6586132725":{"kind":"REWORK","_number":2,"created":"2019-10-01 14:34:56.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/2"}}},"commit":{"parents":[{"commit":"5f16a02137f8fadd4891ff539ff100fa9f358add","subject":"Fix invalid link index","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f16a02137f8fadd4891ff539ff100fa9f358add"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-01 14:34:54.000000000","tz":-300},"subject":"Add \u0027Core Liaison\u0027 section to spec template","message":"Add \u0027Core Liaison\u0027 section to spec template\n\nRequire a spec to be sponsored by a member of the core team.\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- \"Non-core contributors don\u0027t understand how nova dev culture works\"-itis\n- \"Nobody but me cares about my feature\" syndrome\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them\" disorder\n\nIn the current commit, this section is optional in ussuri, but I would\nprefer to amend the existing approved specs to add it, and make it\nrequired immediately.\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ef48a18450ab5fc17aa595008457c6586132725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ef48a18450ab5fc17aa595008457c6586132725"}]},"branch":"refs/heads/master"},"998c332e19413883e561dc2b28e1777680bcefbf":{"kind":"REWORK","_number":3,"created":"2019-10-03 15:09:42.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/3"}}},"commit":{"parents":[{"commit":"5f16a02137f8fadd4891ff539ff100fa9f358add","subject":"Fix invalid link index","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f16a02137f8fadd4891ff539ff100fa9f358add"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-03 15:08:27.000000000","tz":-300},"subject":"Add \u0027Core Liaison\u0027 section to spec template","message":"Add \u0027Core Liaison\u0027 section to spec template\n\nRequire a spec to be sponsored by a member of the core team.\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- \"Non-core contributors don\u0027t understand how nova dev culture works\"-itis\n- \"Nobody but me cares about my feature\" syndrome\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them\" disorder\n\nThe current commit edits already-approved specs to include the new\nsection. Some are filled in, but some are still TBD. These should be\nedited ASAP via subsequent patches.\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/998c332e19413883e561dc2b28e1777680bcefbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/998c332e19413883e561dc2b28e1777680bcefbf"}]},"branch":"refs/heads/master"},"adbb8376b650afd7176563512b9f92aa8449980f":{"kind":"REWORK","_number":4,"created":"2019-10-03 21:19:00.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/4"}}},"commit":{"parents":[{"commit":"5f16a02137f8fadd4891ff539ff100fa9f358add","subject":"Fix invalid link index","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f16a02137f8fadd4891ff539ff100fa9f358add"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-03 21:18:56.000000000","tz":-300},"subject":"Add \u0027Core Liaison\u0027 spec process","message":"Add \u0027Core Liaison\u0027 spec process\n\nAdd a mandatory \u0027Core Liaison\u0027 section to the spec template, requiring\nspecs to be sponsored by an experienced nova developer (not necessarily\nan actual core -- this is explained in the docs).\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- \"Non-core contributors don\u0027t understand how nova dev culture works\"-itis\n- \"Nobody but me cares about my feature\" syndrome\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them\" disorder\n\nThe current commit edits already-approved specs to include the new\nsection.\n\nThe main README is updated with a new section to describe more\nthoroughly what this core liaison thing is. At the same time, since this\nimpacts several aspects of the process, the remainder of the README is\ntweaked and updated accordingly. Of note:\n\n* We\u0027re now going to make distinct use of the launchpad blueprint\u0027s\n  \"Definition\" and \"Direction\" fields. As such, we can still decide to\n  defer a blueprint whose spec is merged in the \u0027approved\u0027 directory.\n  (Which really isn\u0027t different than what we were doing before; it\u0027s\n  just that now we can do it for reasons other than \"oops, this didn\u0027t\n  get finished in time\".)\n* The single-core-approval rule for previously approved specifications\n  is removed [1].\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-10-03.log.html#t2019-10-03T16:08:05\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/adbb8376b650afd7176563512b9f92aa8449980f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/adbb8376b650afd7176563512b9f92aa8449980f"}]},"branch":"refs/heads/master"},"2c14e37afb076eb14d3b58c5bae04adfa747193c":{"kind":"REWORK","_number":5,"created":"2019-10-04 14:32:34.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/5"}}},"commit":{"parents":[{"commit":"fbfb289679267e72903fedb97614b586ac0d79bc","subject":"Add image-precache-support spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbfb289679267e72903fedb97614b586ac0d79bc"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-04 14:31:57.000000000","tz":-300},"subject":"Add \u0027Core Liaison\u0027 spec process","message":"Add \u0027Core Liaison\u0027 spec process\n\nAdd a mandatory \u0027Core Liaison\u0027 section to the spec template, requiring\nspecs to be sponsored by an experienced nova developer (not necessarily\nan actual core -- this is explained in the docs).\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- Non-core contributors don\u0027t understand how nova dev culture works\n- \"Nobody but me cares about my feature\"\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them?\"\n\nThe current commit edits already-approved specs to include the new\nsection.\n\nThe main README is updated with a new section to describe more\nthoroughly what this core liaison thing is. At the same time, since this\nimpacts several aspects of the process, the remainder of the README is\ntweaked and updated accordingly. Of note:\n\n* We\u0027re now going to make distinct use of the launchpad blueprint\u0027s\n  \"Definition\" and \"Direction\" fields. As such, we can still decide to\n  defer a blueprint whose spec is merged in the \u0027approved\u0027 directory.\n  (Which really isn\u0027t different than what we were doing before; it\u0027s\n  just that now we can do it for reasons other than \"oops, this didn\u0027t\n  get finished in time\".)\n* The single-core-approval rule for previously approved specifications\n  is removed [1].\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-10-03.log.html#t2019-10-03T16:08:05\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2c14e37afb076eb14d3b58c5bae04adfa747193c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2c14e37afb076eb14d3b58c5bae04adfa747193c"}]},"branch":"refs/heads/master"},"249466f04306dd34bbd42c21cf86bd754c9ceb99":{"kind":"REWORK","_number":6,"created":"2019-10-07 19:35:32.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/6"}}},"commit":{"parents":[{"commit":"fbfb289679267e72903fedb97614b586ac0d79bc","subject":"Add image-precache-support spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbfb289679267e72903fedb97614b586ac0d79bc"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-07 19:34:40.000000000","tz":-300},"subject":"Add \u0027Feature Liaison\u0027 spec process","message":"Add \u0027Feature Liaison\u0027 spec process\n\nAdd a mandatory \u0027Feature Liaison\u0027 section to the spec template,\nrequiring specs to be sponsored by an experienced nova developer (not\nnecessarily an actual core -- this is explained in the docs).\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- Inexperienced contributors don\u0027t understand how nova dev culture works\n- \"Nobody but me cares about my feature\"\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them?\"\n\nThe current commit edits already-approved specs to include the new\nsection.\n\nThe main README is updated with a new section to describe more\nthoroughly what this feature liaison thing is. At the same time, since\nthis impacts several aspects of the process, the remainder of the README\nis tweaked and updated accordingly. Of note:\n\n* We\u0027re now going to make distinct use of the launchpad blueprint\u0027s\n  \"Definition\" and \"Direction\" fields. As such, we can still decide to\n  defer a blueprint whose spec is merged in the \u0027approved\u0027 directory.\n  (Which really isn\u0027t different than what we were doing before; it\u0027s\n  just that now we can do it for reasons other than \"oops, this didn\u0027t\n  get finished in time\".)\n* The single-core-approval rule for previously approved specifications\n  is removed [1].\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-10-03.log.html#t2019-10-03T16:08:05\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/249466f04306dd34bbd42c21cf86bd754c9ceb99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/249466f04306dd34bbd42c21cf86bd754c9ceb99"}]},"branch":"refs/heads/master"},"ac72442e689d2e25abf097d4a7c20d711142fc04":{"kind":"REWORK","_number":7,"created":"2019-10-14 16:25:55.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/7"}}},"commit":{"parents":[{"commit":"fbfb289679267e72903fedb97614b586ac0d79bc","subject":"Add image-precache-support spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbfb289679267e72903fedb97614b586ac0d79bc"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-14 16:25:49.000000000","tz":-300},"subject":"Add \u0027Feature Liaison\u0027 spec process","message":"Add \u0027Feature Liaison\u0027 spec process\n\nAdd a mandatory \u0027Feature Liaison\u0027 section to the spec template,\nrequiring specs to be sponsored by an experienced nova developer (not\nnecessarily an actual core -- this is explained in the docs).\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- Inexperienced contributors don\u0027t understand how nova dev culture works\n- \"Nobody but me cares about my feature\"\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them?\"\n\nThe current commit edits already-approved specs to include the new\nsection.\n\nThe main README is updated with a new section to describe more\nthoroughly what this feature liaison thing is. At the same time, since\nthis impacts several aspects of the process, the remainder of the README\nis tweaked and updated accordingly. Of note:\n\n* We\u0027re now going to make distinct use of the launchpad blueprint\u0027s\n  \"Definition\" and \"Direction\" fields. As such, we can still decide to\n  defer a blueprint whose spec is merged in the \u0027approved\u0027 directory.\n  (Which really isn\u0027t different than what we were doing before; it\u0027s\n  just that now we can do it for reasons other than \"oops, this didn\u0027t\n  get finished in time\".)\n* The single-core-approval rule for previously approved specifications\n  is removed [1].\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-10-03.log.html#t2019-10-03T16:08:05\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ac72442e689d2e25abf097d4a7c20d711142fc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ac72442e689d2e25abf097d4a7c20d711142fc04"}]},"branch":"refs/heads/master"},"a1421755787a6541a322f7add36f987a2920e882":{"kind":"REWORK","_number":8,"created":"2019-10-14 16:52:15.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/685857/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/57/685857/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/57/685857/8"}}},"commit":{"parents":[{"commit":"51496fd901d5ffd5278e6e27d3b139457d26a6cb","subject":"Merge \"Fix followup comments of policy-defaults-refresh spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/51496fd901d5ffd5278e6e27d3b139457d26a6cb"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-30 22:53:17.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-14 16:52:09.000000000","tz":-300},"subject":"Add \u0027Feature Liaison\u0027 spec process","message":"Add \u0027Feature Liaison\u0027 spec process\n\nAdd a mandatory \u0027Feature Liaison\u0027 section to the spec template,\nrequiring specs to be sponsored by an experienced nova developer (not\nnecessarily an actual core -- this is explained in the docs).\n\nThis in an effort to mitigate several problems noted in previous\ndevelopment cycles:\n- Inexperienced contributors don\u0027t understand how nova dev culture works\n- \"Nobody but me cares about my feature\"\n- \"Whom should I ask for reviews first, especially if I don\u0027t even know\n  whether I\u0027m ready for them?\"\n\nThe current commit edits already-approved specs to include the new\nsection.\n\nThe main README is updated with a new section to describe more\nthoroughly what this feature liaison thing is. At the same time, since\nthis impacts several aspects of the process, the remainder of the README\nis tweaked and updated accordingly. Of note:\n\n* We\u0027re now going to make distinct use of the launchpad blueprint\u0027s\n  \"Definition\" and \"Direction\" fields. As such, we can still decide to\n  defer a blueprint whose spec is merged in the \u0027approved\u0027 directory.\n  (Which really isn\u0027t different than what we were doing before; it\u0027s\n  just that now we can do it for reasons other than \"oops, this didn\u0027t\n  get finished in time\".)\n* The single-core-approval rule for previously approved specifications\n  is removed [1].\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-10-03.log.html#t2019-10-03T16:08:05\n\nChange-Id: Ibb3a2990e23aecf3ea7ebc1b47413396b676f2d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a1421755787a6541a322f7add36f987a2920e882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a1421755787a6541a322f7add36f987a2920e882"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
