)]}'
{"id":"openstack%2Fnova-specs~686058","triplet_id":"openstack%2Fnova-specs~master~I55b9490178dd215eb6e9f7d1f48e28f3eeaec63d","project":"openstack/nova-specs","branch":"master","topic":"bp/policy-defaults-refresh","hashtags":[],"change_id":"I55b9490178dd215eb6e9f7d1f48e28f3eeaec63d","subject":"Re-propose policy-defaults-refresh spec for Ussuri","status":"MERGED","created":"2019-10-01 21:55:01.000000000","updated":"2019-10-28 19:36:13.000000000","submitted":"2019-10-28 19:34:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686058-1572291294041-cc9bba01","meta_rev_id":"896f556235f489d8a32891d50d5ad48debec3ec2","_number":686058,"virtual_id_number":686058,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 19:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2019-10-28 19:12:39.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2019-10-28 19:12:39.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 09:20:07.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-10-28 19:04:46.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-10-28 19:12:39.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-28 19:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5143fcabe0194464b9e888505a930422509676f8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-01 21:55:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8902b19f4f4ccffc0b773bd526c92a7c4ed09eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 22:16:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee895e2b6246448485e2576ca79c57b6 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e76512c6a1423c86680a12a651d113 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"404b4f23ba9458bfd2be52727bc3fe65e08f0d96","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-07 21:55:13.000000000","message":"Patch Set 1:\n\nefried@efried-ThinkPad-W520:~/openstack/nova-specs$ diff specs/train/approved/policy-default-refresh.rst specs/ussuri/approved/\n548a549,550\n\u003e    * - Ussuri\n\u003e      - Re-proposed","accounts_in_message":[],"_revision_number":1},{"id":"8255b23da2310c4fa1b68fe3a88b9eeceb29a6eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-14 16:26:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c675dac42a0ad598da31aeca8ef4f61c6c81a2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 16:42:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0d0a90f76b40a9bf464ec25b2adbd8 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9df2b00464a54028a2df6122fdf4145a : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f4dc7fd355657cabb3f01c47e8005c39d3e20719","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-16 18:01:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8a967a6f792892a2a72973aefd9329af33850bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 18:17:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9dcb97a67c1464faf12de06576ab09e : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b20c96ca5f04b7faddd6268106581ce : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"93fd3bb5192be8ff60b70392f0db0b72aba0c787","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-16 19:01:40.000000000","message":"Patch Set 3:\n\n(Procedural review only.)\n\nDiff at PS1 from the merged Train spec is only the History table (as mentioned previously).\n\nPS1..3 has a couple of clarifications, some formatting tweaks, template compliance update (Feature Liaison added), and removal of references to /os-cells APIs (cellsv1, which died in Train).\n\nTL;DR this is still identical to the Train spec for all intents and purposes afaict.","accounts_in_message":[],"_revision_number":3},{"id":"852f6d7e763c58e55f5d2e303bf04133a0be6104","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-17 11:31:56.000000000","message":"Patch Set 3:\n\n\u003e (Procedural review only.)\n \u003e \n \u003e Diff at PS1 from the merged Train spec is only the History table\n \u003e (as mentioned previously).\n \u003e \n \u003e PS1..3 has a couple of clarifications, some formatting tweaks,\n \u003e template compliance update (Feature Liaison added), and removal of\n \u003e references to /os-cells APIs (cellsv1, which died in Train).\n \u003e \n \u003e TL;DR this is still identical to the Train spec for all intents and\n \u003e purposes afaict.\n\nyeah, basically the followup nit patch got merged - https://review.opendev.org/#/c/669196/","accounts_in_message":[],"_revision_number":3},{"id":"c88ece378b3789d58944422b951c27e6e80e402d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-10-28 09:20:07.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nI still think this is awesome and basically #1 or #2 in Nova feature request from our customers.","accounts_in_message":[],"_revision_number":3},{"id":"ac1c82345214ff279953862333686dcc9bfecf0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-28 17:20:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a5446e288c8b696a6e1b203b91e7ac554f5b1fac","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-10-28 19:04:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThis LGTM too. We have customers who would like to be able to configure policy to allow non-admin to do particular actions such as live migrate and this work would make that possible once finished.","accounts_in_message":[],"_revision_number":3},{"id":"bb9e1321e58e76ce3986974e158c1193d37d901d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:04:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b840299081f6797a49df49e94990840809d42b94","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 19:12:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f73e03b81b9b297d8d94a9bb5673298ea7c1e30e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 19:12:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nI just made the file path match the blueprint name, reapproving.","accounts_in_message":[],"_revision_number":4},{"id":"60a0dfb3bdc6018dce45c40a0ecf71a200f4ecce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:20:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7d6f60813c44b2bf0715f494ad3e17 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0dae5ae818423d9c19d7c84f8b9545 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":4},{"id":"691ed9810183df8e499f4a680bd1ea5828de98ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:20:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8869e9bf6ef42c2c54effa6682079fc646aba437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:34:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c865464d6d1f43bda55a65e0d64a1845 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038ac93ef47c4c54821ee678365c3491 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":4},{"id":"7a52da7a48eeb72f5d18e548187d5db1d7e296e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:34:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c6e2062c9fc9085bb48d6e7080dab08c239d3eda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:36:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/f6e2591a53e243c99b2640f96f4cd6a6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f9d8d34be96d641da5b3a537e7cd6a8da7e0306a","revisions":{"eaa88962f183cc8454997c12dbf230381bb5bcd2":{"kind":"REWORK","_number":1,"created":"2019-10-01 21:55:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/686058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/686058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/1"}}},"commit":{"parents":[{"commit":"5f16a02137f8fadd4891ff539ff100fa9f358add","subject":"Fix invalid link index","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f16a02137f8fadd4891ff539ff100fa9f358add"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 21:53:14.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 21:54:59.000000000","tz":0},"subject":"Re-propose policy-defaults-refresh spec for Ussuri","message":"Re-propose policy-defaults-refresh spec for Ussuri\n\nPreviously-approved: Train\n\nSpec for blueprint policy-defaults-refresh\n\nChange-Id: I55b9490178dd215eb6e9f7d1f48e28f3eeaec63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eaa88962f183cc8454997c12dbf230381bb5bcd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eaa88962f183cc8454997c12dbf230381bb5bcd2"}]},"branch":"refs/heads/master"},"a9342ddc51fcf7e0695b89a4f4eb84d4a72827bf":{"kind":"REWORK","_number":2,"created":"2019-10-14 16:26:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/686058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/686058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/2"}}},"commit":{"parents":[{"commit":"51496fd901d5ffd5278e6e27d3b139457d26a6cb","subject":"Merge \"Fix followup comments of policy-defaults-refresh spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/51496fd901d5ffd5278e6e27d3b139457d26a6cb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-14 16:26:31.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-14 16:26:31.000000000","tz":0},"subject":"Re-propose policy-defaults-refresh spec for Ussuri","message":"Re-propose policy-defaults-refresh spec for Ussuri\n\nPreviously-approved: Train\n\nSpec for blueprint policy-defaults-refresh\n\nChange-Id: I55b9490178dd215eb6e9f7d1f48e28f3eeaec63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a9342ddc51fcf7e0695b89a4f4eb84d4a72827bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a9342ddc51fcf7e0695b89a4f4eb84d4a72827bf"}]},"branch":"refs/heads/master"},"d06910ffdc62648ae2a7569ab5c9d97eec0d6273":{"kind":"REWORK","_number":3,"created":"2019-10-16 18:01:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/686058/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/686058/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/3"}}},"commit":{"parents":[{"commit":"51496fd901d5ffd5278e6e27d3b139457d26a6cb","subject":"Merge \"Fix followup comments of policy-defaults-refresh spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/51496fd901d5ffd5278e6e27d3b139457d26a6cb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-14 16:26:31.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-16 18:01:01.000000000","tz":0},"subject":"Re-propose policy-defaults-refresh spec for Ussuri","message":"Re-propose policy-defaults-refresh spec for Ussuri\n\nPreviously-approved: Train\n\nSpec for blueprint policy-defaults-refresh\n\nChange-Id: I55b9490178dd215eb6e9f7d1f48e28f3eeaec63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d06910ffdc62648ae2a7569ab5c9d97eec0d6273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d06910ffdc62648ae2a7569ab5c9d97eec0d6273"}]},"branch":"refs/heads/master"},"f9d8d34be96d641da5b3a537e7cd6a8da7e0306a":{"kind":"REWORK","_number":4,"created":"2019-10-28 19:12:16.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/58/686058/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/58/686058/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/58/686058/4"}}},"commit":{"parents":[{"commit":"51496fd901d5ffd5278e6e27d3b139457d26a6cb","subject":"Merge \"Fix followup comments of policy-defaults-refresh spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/51496fd901d5ffd5278e6e27d3b139457d26a6cb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-14 16:26:31.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-28 19:12:06.000000000","tz":-300},"subject":"Re-propose policy-defaults-refresh spec for Ussuri","message":"Re-propose policy-defaults-refresh spec for Ussuri\n\nPreviously-approved: Train\n\nSpec for blueprint policy-defaults-refresh\n\nChange-Id: I55b9490178dd215eb6e9f7d1f48e28f3eeaec63d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9d8d34be96d641da5b3a537e7cd6a8da7e0306a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9d8d34be96d641da5b3a537e7cd6a8da7e0306a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
