)]}'
{"id":"openstack%2Fnova-specs~686934","triplet_id":"openstack%2Fnova-specs~master~I416f43b243b0e75118eb22fa16e8c4de5be09d41","project":"openstack/nova-specs","branch":"master","hashtags":[],"change_id":"I416f43b243b0e75118eb22fa16e8c4de5be09d41","subject":"Up load new blueprint snapshot-vms-with-state-of-processes","status":"ABANDONED","created":"2019-10-06 12:21:27.000000000","updated":"2019-10-07 04:05:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"54f76eb25d775736ff602b01a5941ba4827b6e3a","_number":686934,"virtual_id_number":686934,"owner":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-06 12:35:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"all":[{"value":-1,"date":"2019-10-06 18:25:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"all":[{"value":-1,"date":"2019-10-06 18:26:29.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-06 12:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f582eaabc3aca9375b39aa99e199413c33f379d","author":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"date":"2019-10-06 12:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54b9289b077caf046d63fabc081f4367ea9119e2","author":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"date":"2019-10-06 12:22:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c96bfb7834b07ec07a4f31ff06703b92f2344cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 12:35:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe4808464ce453e99934b04ba15fcd6 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30dedc03f2a4758bf29b1e938bdd2d9 : FAILURE in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a3f5660bbcb036dd1b680cf330148ad507b06f29","author":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"date":"2019-10-06 18:25:11.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"acdce50c019e6ae680ec8b8d99dc0fc3b1091351","author":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"date":"2019-10-06 18:26:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"54f76eb25d775736ff602b01a5941ba4827b6e3a","author":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"date":"2019-10-07 04:05:32.000000000","message":"Abandoned\n\nfail","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b264fededb4b94d8a97931caca087fa61293f679","revisions":{"b264fededb4b94d8a97931caca087fa61293f679":{"kind":"REWORK","_number":1,"created":"2019-10-06 12:21:27.000000000","uploader":{"_account_id":30977,"name":"Son Pham","email":"phamngocsonls@gmail.com","username":"SonPham"},"ref":"refs/changes/34/686934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/686934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/686934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/686934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/686934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/686934/1"}}},"commit":{"parents":[{"commit":"fbfb289679267e72903fedb97614b586ac0d79bc","subject":"Add image-precache-support spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbfb289679267e72903fedb97614b586ac0d79bc"}]}],"author":{"name":"SonPham","email":"phamngocsonls@gmail.com","date":"2019-10-06 12:20:39.000000000","tz":420},"committer":{"name":"SonPham","email":"phamngocsonls@gmail.com","date":"2019-10-06 12:21:24.000000000","tz":420},"subject":"Up load new blueprint snapshot-vms-with-state-of-processes","message":"Up load new blueprint snapshot-vms-with-state-of-processes\n\nChange-Id: I416f43b243b0e75118eb22fa16e8c4de5be09d41\nSigned-off-by: SonPham \u003cphamngocsonls@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b264fededb4b94d8a97931caca087fa61293f679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b264fededb4b94d8a97931caca087fa61293f679"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
