)]}'
{"id":"openstack%2Fnova-specs~687199","triplet_id":"openstack%2Fnova-specs~master~Id0b7927d6a6cd4a3a41b44f1de160ea117e0cfbc","project":"openstack/nova-specs","branch":"master","topic":"bp/expose-auto-converge-post-copy","hashtags":[],"change_id":"Id0b7927d6a6cd4a3a41b44f1de160ea117e0cfbc","subject":"Re-propose expose-auto-converge-post-copy spec for Ussuri","status":"ABANDONED","created":"2019-10-08 02:43:33.000000000","updated":"2019-11-14 09:12:02.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8d5d6dd95056941f792efaffd049792f48572ef5","_number":687199,"virtual_id_number":687199,"owner":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-08 01:31:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-11-12 15:45:53.000000000","_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"date":"2019-11-11 00:03:48.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 07:32:42.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-10-10 07:33:52.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-10-10 07:34:10.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-10-30 17:15:04.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-06 02:29:02.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2019-11-08 01:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 00:03:48.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-11-12 15:45:53.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"}],"messages":[{"id":"4e340d190e938f0c0067e430060f635151ede038","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-10-08 02:43:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ebfe3873a5c9210987495f2648f1b0bdbae257b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 02:52:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0b9ed836574d669b5adb038468754b : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828586f8d67e4256aae9dc15b1130a79 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"372fdb0f0c61b1a28f264a37085c968faeac8c9f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-30 17:15:04.000000000","message":"Patch Set 1: Code-Review-1\n\nprocedural downvote; this is going to need a \"Feature Liaison\" as described here http://specs.openstack.org/openstack/nova-specs/readme.html#feature-liaison-faq","accounts_in_message":[],"_revision_number":1},{"id":"f70fc9410ebd70f0a01d9514417585269a2f058f","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-11-05 12:50:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33e866c82a1eeb830e413eca3ea2a8fdcf28b040","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-11-05 12:51:05.000000000","message":"Patch Set 2:\n\n\u003e procedural downvote; this is going to need a \"Feature Liaison\" as\n \u003e described here http://specs.openstack.org/openstack/nova-specs/readme.html#feature-liaison-faq\n\ndone","accounts_in_message":[],"_revision_number":2},{"id":"716fdf8d367b3a7650174d7ee26f8a34fdbeeb22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:00:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f574e8efc34d768147a6f6b21e9768 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b50a55765343c4bdb9e4a1c1146afb : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"6e5d4534209d155976b7dc1528c27c6da79644d8","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-11-06 02:29:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"af2e2874bcda9f544797a20e414ecfa88f061c46","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-11-07 15:25:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87897d2a0ae2829c9711b6a61c25f4fd9b67f82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 15:40:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0c4b7e206d497a86361014cec86ab8 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2c1987731b4346986e870fbe8871ba : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c5452ad70a2b91f60ebe2ee1e70b159b809ddd34","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-11-08 01:23:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2579c563cd743b2a5442a3c4db0488a0b06713f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 01:31:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20e0f8b6e264f45933a7262dc89cc92 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0cffbe17c5b4b518abbf0a37c9f54e9 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"8f53c103ccfa14a1d706ad0c7da365388d812639","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-11-14 09:12:02.000000000","message":"Abandoned\n\nAbandoned, another method has been proposed to solve the problem. [1]_\n\nRef:\n..[1] https://review.opendev.org/#/c/693655/","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"79b2c4d17ebc2b7dc0117e70027f133adbd2edc5","revisions":{"bdcd3af71593e72e343615ac3846c1e8fcd794d8":{"kind":"REWORK","_number":1,"created":"2019-10-08 02:43:33.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/99/687199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/99/687199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/1"}}},"commit":{"parents":[{"commit":"fbfb289679267e72903fedb97614b586ac0d79bc","subject":"Add image-precache-support spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fbfb289679267e72903fedb97614b586ac0d79bc"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-10-08 02:33:24.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-10-08 02:42:55.000000000","tz":480},"subject":"Re-propose expose-auto-converge-post-copy spec for Ussuri","message":"Re-propose expose-auto-converge-post-copy spec for Ussuri\n\nNo changes from Train version except History.\n\nChange-Id: Id0b7927d6a6cd4a3a41b44f1de160ea117e0cfbc\nPreviously-approved: Train\nBlueprint: expose-auto-converge-post-copy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bdcd3af71593e72e343615ac3846c1e8fcd794d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bdcd3af71593e72e343615ac3846c1e8fcd794d8"}]},"branch":"refs/heads/master"},"606f9550052f22aa26c573e78bc7061b308cf7b1":{"kind":"REWORK","_number":2,"created":"2019-11-05 12:50:39.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/99/687199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/99/687199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/2"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-10-08 02:33:24.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-11-05 12:50:17.000000000","tz":480},"subject":"Re-propose expose-auto-converge-post-copy spec for Ussuri","message":"Re-propose expose-auto-converge-post-copy spec for Ussuri\n\nNo changes from Train version except History.\n\nChange-Id: Id0b7927d6a6cd4a3a41b44f1de160ea117e0cfbc\nPreviously-approved: Train\nBlueprint: expose-auto-converge-post-copy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/606f9550052f22aa26c573e78bc7061b308cf7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/606f9550052f22aa26c573e78bc7061b308cf7b1"}]},"branch":"refs/heads/master"},"81c231c8d755757276e683c6350408dfc5af3f9e":{"kind":"REWORK","_number":3,"created":"2019-11-07 15:25:52.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/99/687199/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/99/687199/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/3"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-10-08 02:33:24.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-11-07 15:24:53.000000000","tz":480},"subject":"Re-propose expose-auto-converge-post-copy spec for Ussuri","message":"Re-propose expose-auto-converge-post-copy spec for Ussuri\n\nChange the problem description and use case.\n\nChange-Id: Id0b7927d6a6cd4a3a41b44f1de160ea117e0cfbc\nPreviously-approved: Train\nBlueprint: expose-auto-converge-post-copy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/81c231c8d755757276e683c6350408dfc5af3f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/81c231c8d755757276e683c6350408dfc5af3f9e"}]},"branch":"refs/heads/master"},"79b2c4d17ebc2b7dc0117e70027f133adbd2edc5":{"kind":"REWORK","_number":4,"created":"2019-11-08 01:23:31.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/99/687199/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/99/687199/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/99/687199/4"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-10-08 02:33:24.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-11-08 01:22:45.000000000","tz":480},"subject":"Re-propose expose-auto-converge-post-copy spec for Ussuri","message":"Re-propose expose-auto-converge-post-copy spec for Ussuri\n\nChange the problem description and use case.\nAdd a new alternatives.\n\nChange-Id: Id0b7927d6a6cd4a3a41b44f1de160ea117e0cfbc\nPreviously-approved: Train\nBlueprint: expose-auto-converge-post-copy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79b2c4d17ebc2b7dc0117e70027f133adbd2edc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79b2c4d17ebc2b7dc0117e70027f133adbd2edc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
