)]}'
{"id":"openstack%2Fnova-specs~693226","triplet_id":"openstack%2Fnova-specs~master~I8697bf3ca72c94c55be0597441921d39fe6f2d70","project":"openstack/nova-specs","branch":"master","topic":"bp/nova-audit","hashtags":[],"change_id":"I8697bf3ca72c94c55be0597441921d39fe6f2d70","subject":"Add nova-audit spec","status":"MERGED","created":"2019-11-06 17:43:22.000000000","updated":"2020-02-14 10:13:55.000000000","submitted":"2020-02-14 10:11:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":58,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693226-1581675097625-86a07e3f","meta_rev_id":"f47ea93cfa8d48dbd83d0f28c770336e6b1ba883","_number":693226,"virtual_id_number":693226,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-11-07 18:54:21.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-11-27 03:41:41.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 10:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},"all":[{"value":1,"date":"2020-01-08 07:44:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":1,"date":"2020-01-31 08:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"date":"2020-02-13 19:02:02.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-11-27 00:23:54.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-11 19:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-11-27 02:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},{"value":2,"date":"2020-02-14 09:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2020-02-14 09:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-01-30 23:02:32.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2020-02-14 09:55:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 19:14:05.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-27 00:23:54.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-11-27 02:29:48.000000000","updated_by":{"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},"reviewer":{"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},"state":"REVIEWER"},{"updated":"2019-11-27 03:41:41.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-01-08 07:44:47.000000000","updated_by":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"reviewer":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"state":"REVIEWER"},{"updated":"2020-01-30 23:02:32.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-31 08:14:54.000000000","updated_by":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"reviewer":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"state":"REVIEWER"},{"updated":"2020-02-13 19:02:02.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-02-14 09:04:04.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2020-02-14 09:55:34.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-02-14 10:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11c4a340c67d322e98e6bea68420a34b844a2e58","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 17:43:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfc80ce2aadbd6392212ab3231ae2968576d0b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 17:53:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9ff67eeded4e279608dfae6a2c935a : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962757b687d04dac97fede5636a8aea4 : FAILURE in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f5fd392eb9d96c4c651c1f8fc1616a94bbdecbf6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 18:00:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e4c54d6a30f843e05b0472c1830a5c9bec6b009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 18:07:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/056f4c482f684ac69ded74fd83286875 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcfbbcecfb34c5288be6ea53584a915 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"27daec26d361199b08900de69e309f9dcea0c246","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-06 18:13:08.000000000","message":"Patch Set 1:\n\n(14 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d3697a27852485eaa4b251261ecab223c7902cf8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 18:22:20.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6fc36fab94c17c94c0167f46e2cff6caf04a922c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-06 18:25:53.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nOverall I really like this idea. I agree that we  have a number of periodic tasks/healing activities that are \"central\" and don\u0027t really fit in any one service today. And I have heard throughout the years from operators that it would be ideal if nova could automatically heal itself in any situation where it\u0027s possible. It would save them a lot of time and effort.","accounts_in_message":[],"_revision_number":2},{"id":"579a7da42cb8c6d0bfc4cca6fc73b01e08d9acf5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 18:38:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c600abc7024089bde29b05a108042041b16c61f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 18:41:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22dff5668a22b32da966227b438231b90f2140a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 18:44:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d5a8417f544f148d3840e153ec01b5 : FAILURE in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1fb07d34f34ddc9ebf8dce136527c8 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a46c9b8ce02822b5904be5427186301818881843","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 18:46:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a1a1a9e68bff7320103197342acfc242dff6028e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-06 18:49:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffd67e2f35d0b8abe22b97dbec1cb5691b2cc6bd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-06 18:52:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0679ae208b4b81b3cead840b90281c780487ce40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 18:55:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc490491c54412ba35b0f8207bfa0ef : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f02e088390284d128647f7d781ba31a4 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":4},{"id":"782bdb62f93e9636745cc6f62e5a7d04828d5f96","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-06 18:56:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c6195f56017180b1d7f0f5605b1577e0abb7025","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-07 13:02:04.000000000","message":"Patch Set 4:\n\nAdding cdent because he\u0027s mentioned a few times the need/desire for what the ResourceTracker used to do regarding automatically writing allocations. The heal_allocations command is obviously different from that since the RT would overwrite allocations created by the scheduler (starting in Pike) which was a split brain problem for migrations, which is why we then did migration-based allocations in Queens. Having something periodically running heal_allocations that the operator doesn\u0027t have to setup an explicit cron for though should help in the cases that allocations are lost for an instance and we just want to automatically heal them back based on the flavor (and ports if there are qos ports attached with resource allocation requests).","accounts_in_message":[],"_revision_number":4},{"id":"2b0498e04cfef0cdd185dc6bd5cb141e85d08a4b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-07 13:20:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)\n\nI\u0027m cool with this except for the part about deprecating and moving the existing nova-manage commands. More detailed thoughts within.\n\nAlso, I think this could likely move forward without that if it becomes a sticking point.","accounts_in_message":[],"_revision_number":4},{"id":"799a81bfdc5a1c3fc0249f265dd5ddb885655656","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-07 13:21:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2080e8442cfa1ba059d960b13392bd518c4d470f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-11-07 13:43:13.000000000","message":"Patch Set 4:\n\n(5 comments)\n\nOverall I think this is a great idea. I think are details on the architecture and implementation that can be debated, but the general idea of \"fixing stuff that has to be fixed, that we know how to fix, automatically\" is really good.\n\nOne caveat on that is we\u0027ll want to be conscious about not allowing the presence of this fixer to let things that break be sloppy. The justification of \"it\u0027s a distributed system\" is true, but at least some of those common breakages are the result of bugs that we don\u0027t want to always mask. So it might makes sense that the tool report on the perceived health: \"I had to do task X for N instances Z times in the last M days, you\u0027re sick!\"\n\nI also wonder whether we shouldn\u0027t do more to allow the nova-compute process correct placement on its own. I guess it can\u0027t be fully authoritative (in isolation) about the state of migrations, so that\u0027s perhaps a non-starter, but I\u0027ve always felt a bit awkward about the need to construct a representation of reality from several physical places.\n\nI guess that\u0027s why it\u0027s called virtualization...\n\nAnyway: yes, please do this.","accounts_in_message":[],"_revision_number":4},{"id":"6a3f8053896459a4481384d274502f7ade9cfde9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-07 14:59:44.000000000","message":"Patch Set 4:\n\n(4 comments)\n\n\u003e One caveat on that is we\u0027ll want to be conscious about not allowing\n \u003e the presence of this fixer to let things that break be sloppy. The\n \u003e justification of \"it\u0027s a distributed system\" is true, but at least\n \u003e some of those common breakages are the result of bugs that we don\u0027t\n \u003e want to always mask. So it might makes sense that the tool report\n \u003e on the perceived health: \"I had to do task X for N instances Z\n \u003e times in the last M days, you\u0027re sick!\"\n\nFor sure. I think that this can log all kinds of things at info level, but any healing gets logged as ERROR or WARNING. Operators would, if they run this as a daemon, digest those logs as well and not ignore ERROR any more than they normally would.\n\n \u003e I also wonder whether we shouldn\u0027t do more to allow the\n \u003e nova-compute process correct placement on its own. I guess it can\u0027t\n \u003e be fully authoritative (in isolation) about the state of\n \u003e migrations, so that\u0027s perhaps a non-starter, but I\u0027ve always felt a\n \u003e bit awkward about the need to construct a representation of reality\n \u003e from several physical places.\n\nWell, that\u0027s a different topic somewhat, as (a) many of the things here are not placement related, definitely not suitable for computes to be doing, and not really something you want all conductors doing all the time. And (b) as you say, computes healing with their isolated view of the world leads to other problems which is why we\u0027ve tended away from that activity lately. It seems like the right thing to do, but there are devilish details.","accounts_in_message":[],"_revision_number":4},{"id":"3c2108612e6a9abe1bf43a711896f422230acc20","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-07 15:23:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2b7afbc9cf98436a46555d40f806656c9e5cb25b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-07 15:50:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"512d64a5ef84ffcdb0f3fd06e0d764dd996e031f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-07 18:06:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29cd68693e029a52a5f479c0ee92884f2dc22877","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-07 18:07:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1380f64bb7cfa75c4b37b513c3f7afdc5d02f74","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-07 18:31:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d06ac6dcf94b9755763e99c907cb046734ae8cb8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-07 18:46:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a23bdcf5efc12092dce496c8c3837bf83d57a2e2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-07 18:54:21.000000000","message":"Patch Set 4:\n\nrecheck wth","accounts_in_message":[],"_revision_number":4},{"id":"b8e7d58dc7bc59dcdee4128726286251baa529b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 19:02:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc43ef9093d4bb4b024b65214a46c92 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c7ceedea59469ea08307b94db974da : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"da716573361415db4decc88448240fe5de89d5e4","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-07 19:06:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef24e292eb794b4826c92fc804685cf8d353977f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-08 15:15:30.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c1e3b2ba10800e80b0fd25a457f2db42f764c4e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-11 19:14:05.000000000","message":"Patch Set 4: Code-Review+1\n\nI like the general motive and concept.  This is useful for deployments.","accounts_in_message":[],"_revision_number":4},{"id":"664aa102527c88831d028de03805fa4361aa3c48","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-11 20:13:24.000000000","message":"Patch Set 4:\n\n\u003e I like the general motive and concept.  This is useful for deployments.\n\nHow do you feel about deprecating the nova-manage commands that would be used by the audit command and eventually moving those to the audit command? Would you want that or not care?","accounts_in_message":[],"_revision_number":4},{"id":"26f2e158eeb9513bb5480aa2000e4458f989c3b2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-11-20 17:40:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"c6bcd5db4a860aa3417d78ca3056395d2bae64ae","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-23 21:14:25.000000000","message":"Patch Set 4:\n\n\u003e \u003e I like the general motive and concept.  This is useful for\n \u003e deployments.\n \u003e \n \u003e How do you feel about deprecating the nova-manage commands that\n \u003e would be used by the audit command and eventually moving those to\n \u003e the audit command? Would you want that or not care?\n\nI\u0027d prefer a service that just keeps stuff running properly all the time.","accounts_in_message":[],"_revision_number":4},{"id":"39792f9f8cdce1da338e867dae8e26c079d3adc2","author":{"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},"date":"2019-11-25 20:40:00.000000000","message":"Patch Set 4:\n\n\u003e \u003e I like the general motive and concept.  This is useful for\n \u003e deployments.\n \u003e \n \u003e How do you feel about deprecating the nova-manage commands that\n \u003e would be used by the audit command and eventually moving those to\n \u003e the audit command? Would you want that or not care?\n\nDeprecating nova-manage makes sense to me. I\u0027d rather have this in a service repairing everything on the fly.","accounts_in_message":[],"_revision_number":4},{"id":"43cdd200034e889204e996fc1fe0770236d4a4ce","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-25 21:58:33.000000000","message":"Patch Set 4:\n\n\u003e Deprecating nova-manage makes sense to me. I\u0027d rather have this in a service repairing everything on the fly.\n\nThe spec isn\u0027t proposing that certain nova-manage commands would just be deprecated and eventually removed and replaced by this nova-audit service. It\u0027s saying the nova-manage commands in question would be deprecated and moved to the nova-audit command which could also be run in daemon mode, but the commands themselves wouldn\u0027t be removed so you still have a way to run them on-demand as needed.","accounts_in_message":[],"_revision_number":4},{"id":"76c27d12901c40e1eb375098935cb0115be6964b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-26 22:43:36.000000000","message":"Restored\n\nRestoring as a multi-person effort.","accounts_in_message":[],"_revision_number":4},{"id":"d04a3423d1751748ebe4e7bd2505b5ce25534c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 22:55:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6c2a3106dd471a8052f42d2599b764 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/219507234bbe4eeb89ef6d956a9336b0 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":4},{"id":"71c4fa6238e0673c443d25b81d9c60b6871c9194","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-26 22:57:35.000000000","message":"Patch Set 4: Code-Review+2\n\nOK, I emailed a set of operators for a sanity check with Dan and Matt cc\u0027ed and got responses from eandersson and mnaser expressing a +1 for the idea and no qualms with deprecating/removing the nova-manage commands and moving them to the nova-audit command. And we also have a comment of support from penick on here (posted while in Abandoned state, so +1 wasn\u0027t available) and again no qualms with moving the commands out of nova-manage and into nova-audit.\n\nI have been +2 on this idea and think it\u0027s the right thing to do. I think there is significant need for automatic healing/cleaning functionality to be available in nova and a cleanup of the nova-manage alongside it makes sense. Nova-manage has been a dumping ground for lots of commands and I think it would be beneficial to move the healing/cleaning ones out into their own proper home.\n\nMy only personal hesitation with this proposal was with operators potentially balking at the idea of moving the nova-manage commands. Based on the operator responses to my email, I now feel comfortable that operators will not mind the move aspect of it.\n\nWith the command deprecation and move confirmed not to be a problem with operators, I vote +2.","accounts_in_message":[],"_revision_number":4},{"id":"94afe1ec21f56251fefb541f6df9042e4efec008","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-27 00:16:09.000000000","message":"Patch Set 4:\n\n\u003e Based on the operator responses to my email, I now feel comfortable that operators will not mind the move aspect of it.\n\nFirst, it sucks it was a private email since that cannot be linked here for those not on the email.\n\nI also never received a reply to my last question to eandersson and mnaser in the email thread about the actual deprecation and moving of the commands from nova-manage to nova-audit being something they care about. The responses in the email thread where not nuanced. They were basically, \"yeah I like this idea, sure make it happen\" with no response about my primary concern which is that I think this could be done in two stages: (1) add the new meta command and service which runs the periodics and (2) after that, gauge interest in deprecating and moving the nova-manage commands into a nova-audit command. That\u0027s my major hangup on this and why I\u0027m likely not to +2 it in its current form.","accounts_in_message":[],"_revision_number":4},{"id":"995a2d45ff76ea8737f78aa4306e52061d22e889","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-27 00:17:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f7648758348c9748fcebf3733732b787204ab49","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-27 00:23:54.000000000","message":"Patch Set 4:\n\n\u003e Restoring as a multi-person effort.\n\nAnd if this is a multi-person effort now, and you\u0027re going to be co-author, shouldn\u0027t you abstain from +2ing what would effectively be, at least in part, your own spec?","accounts_in_message":[],"_revision_number":4},{"id":"ae5f1f7d2f135bbf035f939c2b19c7287fe5521d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-27 00:32:02.000000000","message":"Patch Set 4:\n\n\u003e \u003e Based on the operator responses to my email, I now feel\n \u003e comfortable that operators will not mind the move aspect of it.\n \u003e \n \u003e First, it sucks it was a private email since that cannot be linked\n \u003e here for those not on the email.\n\nIt does, but no one replied to the mail to the public list [1] earlier so I thought I\u0027d try a different avenue. I sent that email for myself because I was overthinking it and worried if operators would be upset about commands moving. I didn\u0027t want to spam everyone for a personal sanity check.\n\nI was +2 save for that specific worry and my worry was alleviated by me asking for a sanity check directly.\n\n \u003e I also never received a reply to my last question to eandersson and\n \u003e mnaser in the email thread about the actual deprecation and moving\n \u003e of the commands from nova-manage to nova-audit being something they\n \u003e care about. The responses in the email thread where not nuanced.\n \u003e They were basically, \"yeah I like this idea, sure make it happen\"\n \u003e with no response about my primary concern which is that I think\n \u003e this could be done in two stages: (1) add the new meta command and\n \u003e service which runs the periodics and (2) after that, gauge interest\n \u003e in deprecating and moving the nova-manage commands into a\n \u003e nova-audit command. That\u0027s my major hangup on this and why I\u0027m\n \u003e likely not to +2 it in its current form.\n\nI understand. I thought about it more and ended up agreeing with Dan that it should be done together as a package deal. I think it would be less confusing and less ugly to leave things separated in an interim step. Just MHO.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-November/010595.html","accounts_in_message":[],"_revision_number":4},{"id":"8bd3be26967d50cc35d86fa941f8bd8a4700fcf6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-27 00:39:43.000000000","message":"Patch Set 4: Code-Review+1\n\n\u003e \u003e Restoring as a multi-person effort.\n \u003e \n \u003e And if this is a multi-person effort now, and you\u0027re going to be\n \u003e co-author, shouldn\u0027t you abstain from +2ing what would effectively\n \u003e be, at least in part, your own spec?\n\nI wasn\u0027t thinking about it that way because this was not my idea, but I see your point if I am going to have to help with this. I\u0027ll downgrade to a +1 and upgrade back to +2 if Dan decides he\u0027s willing to hammer out the entire thing during the holiday lull.","accounts_in_message":[],"_revision_number":4},{"id":"033a7589cf31ea7e2ba94e0e380a512cae782989","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-27 00:40:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9b8f2e3af582964485a764b6664b7b4b00cc643","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-27 00:52:17.000000000","message":"Patch Set 4:\n\n\u003e I understand. I thought about it more and ended up agreeing with\n \u003e Dan that it should be done together as a package deal. I think it\n \u003e would be less confusing and less ugly to leave things separated in\n \u003e an interim step. Just MHO.\n\nSigh,\n\n*less confusing and less ugly *THAN* to leave things separated in an interim step","accounts_in_message":[],"_revision_number":4},{"id":"bc1358279810eff4ce8ac2d8565c85a2fb45ea96","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-27 01:36:11.000000000","message":"Patch Set 4:\n\n\u003e \u003e Based on the operator responses to my email, I now feel\n \u003e comfortable that operators will not mind the move aspect of it.\n \u003e \n \u003e First, it sucks it was a private email since that cannot be linked\n \u003e here for those not on the email.\n \u003e \n \u003e I also never received a reply to my last question to eandersson and\n \u003e mnaser in the email thread about the actual deprecation and moving\n \u003e of the commands from nova-manage to nova-audit being something they\n \u003e care about. The responses in the email thread where not nuanced.\n \u003e They were basically, \"yeah I like this idea, sure make it happen\"\n \u003e with no response about my primary concern which is that I think\n \u003e this could be done in two stages: (1) add the new meta command and\n \u003e service which runs the periodics and (2) after that, gauge interest\n \u003e in deprecating and moving the nova-manage commands into a\n \u003e nova-audit command. That\u0027s my major hangup on this and why I\u0027m\n \u003e likely not to +2 it in its current form.\n\nI agree with the meta service and command.  I agree with the idea that this could be an always-running service *or* a one-time thing (depending on how you execute it).\n\nnova-manage being the entry point to that codebase, or nova-audit doesn\u0027t matter to me personally as a deployer.  It would a trivial thing to change the call to the command\n\nPersonally, I\u0027d skip with all of those commands and go to running the service with the \"fix things pls\" mode on.","accounts_in_message":[],"_revision_number":4},{"id":"cef3c671851fd9435db6d49fb75fee71d1989dbf","author":{"_account_id":9452,"name":"James Penick","email":"penick@verizonmedia.com","username":"epim"},"date":"2019-11-27 02:29:48.000000000","message":"Patch Set 4: Code-Review+1\n\n\u003e \u003e Deprecating nova-manage makes sense to me. I\u0027d rather have this\n \u003e in a service repairing everything on the fly.\n \u003e \n \u003e The spec isn\u0027t proposing that certain nova-manage commands would\n \u003e just be deprecated and eventually removed and replaced by this\n \u003e nova-audit service. It\u0027s saying the nova-manage commands in\n \u003e question would be deprecated and moved to the nova-audit command\n \u003e which could also be run in daemon mode, but the commands themselves\n \u003e wouldn\u0027t be removed so you still have a way to run them on-demand\n \u003e as needed.\n\nUnderstood, what I had intended to say was \"deprecating *these from* nova-manage [...] i\u0027d rather have *these* running in a service repairing everything on the fly. \n\nI think that in general we\u0027d be happy to just run this with \u0027fix the problems so things work\u0027 mode enabled. To Mohammed\u0027s point, changing our wrapper scripts from nova-manage to nova-audit would be trivial.","accounts_in_message":[],"_revision_number":4},{"id":"ffe1b7587a436fea1b8fd342d5f68dd04d7c7af3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-11-27 03:41:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6578596fc2ea4a71f6a9e06505fb8c460f170948","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-01-08 07:44:47.000000000","message":"Patch Set 4: Code-Review+1\n\nAs an operator, I too would love a daemon that fixes things in the background, and an interactive mode where it tells me things it can\u0027t fix. (like fsck)\n\nSome questions:\n\n1) How will this work for cross project inconsistency - cinder bdm comes to mind. Also if placement is spun out from nova?\n\n2) Any possibility that the design can accept plugins? It will be useful for custom audits and sharing between ops (e.g. something that only happens for a small number of clouds, but affect a large number of instances so writing code for it is necessary but might not get accepted upstream)\n\nThanks for taking this up; this is a big challenge.","accounts_in_message":[],"_revision_number":4},{"id":"c216583750ebe7d24e41c4f90b3c3b7bd6bc0304","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-30 23:02:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f3f536c203350df37d2e683f92d713daa7129ae7","author":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"date":"2020-01-31 08:14:54.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cbc903a0a0c70f3743171a7fd600f805e2283771","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-13 18:41:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8808ba7932ec684451fd224178b67ca85f6a2fcb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-13 19:02:02.000000000","message":"Patch Set 4: -Code-Review\n\nRemoving my vote since I would be working on this if it gets approved.\n\nI\u0027m also avoiding updating the commit message because I don\u0027t want to lose all of the +1 votes from operators.","accounts_in_message":[],"_revision_number":4},{"id":"297652427267fffd7a9b7b1cd27400e3a374c960","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-02-14 09:04:04.000000000","message":"Patch Set 4: Code-Review+2\n\nSo my +2 here is for the concept. Sounds brillant. Agreed the deamon sounds like an easy way to make sure we throttle things appropriately.\n\nI think deprecating the crazy long list of \"do thing\" nova-manage stuff sounds like a decent tidy up, and given all the +1 form the folks who keep us honest about that kind of thing, then cool.\n\nWhile the detail of all the things we are adding seems sound, I clearly haven\u0027t spent quite a long reviewing all that, but it feels like code review is a better place to argue about those details, and it become clear what the right way forward really is for each of those things.","accounts_in_message":[],"_revision_number":4},{"id":"85920c665731ee2885a6974a5d296853628481fb","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-14 09:55:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(4 comments)\n\nI honestly feel that the path for implementing this spec is now clear. We could still have some open questions about details but I\u0027m all good deferring those detail discussions during the implementation phase since I don\u0027t think it would be more than just some implementation concerns.","accounts_in_message":[],"_revision_number":4},{"id":"7707677f945db7885c4a05556bbc9aef63b6b80d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 09:55:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"69b32e8d2389fec14f7831175e96cfd1f90271e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:11:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b0d00d50b25fe78fa3705efb9000e71785cb25e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:11:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136c93d24ffe4371a3fdb6347750b900 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10bf6bd86366411eadb7dea019f7c68c : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":4},{"id":"bee83d149299444c030d945df95486c8d0e52ab5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:13:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/596a8ed72d0e4ab29cb6e7b02c0c3027 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d649a3544bc386f5d71428d652750cc3c41e11aa","revisions":{"0e9d9ddb8d7c0ff433a4a91da5d86edc298f4c5f":{"kind":"REWORK","_number":1,"created":"2019-11-06 17:43:22.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/693226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/26/693226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/1"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 17:43:13.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 17:43:13.000000000","tz":-480},"subject":"Add nova-audit spec","message":"Add nova-audit spec\n\nChange-Id: I8697bf3ca72c94c55be0597441921d39fe6f2d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0e9d9ddb8d7c0ff433a4a91da5d86edc298f4c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0e9d9ddb8d7c0ff433a4a91da5d86edc298f4c5f"}]},"branch":"refs/heads/master"},"7e0cf21bde5b7c1864a46806aaaeb6234c4d9997":{"kind":"REWORK","_number":2,"created":"2019-11-06 18:00:09.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/693226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/26/693226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/2"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 17:43:13.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 18:00:05.000000000","tz":-480},"subject":"Add nova-audit spec","message":"Add nova-audit spec\n\nChange-Id: I8697bf3ca72c94c55be0597441921d39fe6f2d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e0cf21bde5b7c1864a46806aaaeb6234c4d9997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e0cf21bde5b7c1864a46806aaaeb6234c4d9997"}]},"branch":"refs/heads/master"},"20ba254f32ff7031d58725eec4c519de7f1f93fb":{"kind":"REWORK","_number":3,"created":"2019-11-06 18:38:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/693226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/26/693226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/3"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 17:43:13.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 18:38:40.000000000","tz":-480},"subject":"Add nova-audit spec","message":"Add nova-audit spec\n\nChange-Id: I8697bf3ca72c94c55be0597441921d39fe6f2d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20ba254f32ff7031d58725eec4c519de7f1f93fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20ba254f32ff7031d58725eec4c519de7f1f93fb"}]},"branch":"refs/heads/master"},"d649a3544bc386f5d71428d652750cc3c41e11aa":{"kind":"REWORK","_number":4,"created":"2019-11-06 18:46:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/693226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/26/693226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/26/693226/4"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 17:43:13.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-11-06 18:46:20.000000000","tz":-480},"subject":"Add nova-audit spec","message":"Add nova-audit spec\n\nChange-Id: I8697bf3ca72c94c55be0597441921d39fe6f2d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d649a3544bc386f5d71428d652750cc3c41e11aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d649a3544bc386f5d71428d652750cc3c41e11aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
