)]}'
{"id":"openstack%2Fnova-specs~693414","triplet_id":"openstack%2Fnova-specs~master~If64509785675442ed9e8e8e452bf7dac837c1953","project":"openstack/nova-specs","branch":"master","topic":"bp/provider-config-file","hashtags":[],"change_id":"If64509785675442ed9e8e8e452bf7dac837c1953","subject":"Update provider config spec for identification conflicts","status":"MERGED","created":"2019-11-07 19:33:10.000000000","updated":"2020-01-13 20:57:48.000000000","submitted":"2020-01-13 20:55:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693414-1578948953627-a981c286","meta_rev_id":"417eaf59b1e897188dcf849228e68d1c23e4279e","_number":693414,"virtual_id_number":693414,"owner":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-21 18:17:37.000000000","_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-11-22 00:08:35.000000000","_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-13 20:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":2,"date":"2019-11-21 00:11:40.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},{"value":2,"date":"2020-01-13 20:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},{"value":1,"date":"2020-01-13 20:39:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},{"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 15:13:17.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-11-08 21:15:11.000000000","updated_by":{"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},"reviewer":{"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},"state":"REVIEWER"},{"updated":"2019-11-21 00:11:40.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-22 00:08:35.000000000","updated_by":{"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},"reviewer":{"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},"state":"REVIEWER"},{"updated":"2020-01-13 20:39:42.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-01-13 20:55:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67662f1c9472046325ee8787ef37b6f192e36cad","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-11-07 19:33:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"436a6a095b6d9c4fcfdd539f439b1f1f114336f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 19:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd76114f183740718bb11d4fcb463301 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a51ca18e7104456b33a3f631903be24 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4b74e138413fdfc21f57b98e9305a9aad0a7f2c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-08 15:14:28.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nHm. This might be the right thing. But I want to put this alternative out there for consideration (and noodle it myself for a bit):\n\nIt might be more \"intuitive\" if $COMPUTE_NODE was always treated as the default, and any explicit name/uuid takes precedence no matter where it falls in the sort order.\n\nThen we still have to deal with conflicts of name+uuid, which could be done as first-seen like you\u0027re suggesting here.\n\n...which is more complex to document, so I\u0027m really not sure if it\u0027s better. I need to noodle it out, and would like to hear feedback from others.\n\nBy the way, this gets a lot easier if we rip out the `name` identification method entirely. Then we can let UUIDs take precedence over $COMPUTE_NODE, and just 100% bounce early if we find a duplicate UUID. So -- remind me why we insisted we needed to support `name`? I want to say it was gibi who was pushing for that. (I\u0027ll go paw through the original spec review later if nobody remembers off the top.)","accounts_in_message":[],"_revision_number":1},{"id":"6f94c5c611a2d16868f0001a359347c1e1de5616","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-11-09 01:02:35.000000000","message":"Patch Set 1:\n\n\u003e It might be more \"intuitive\" if $COMPUTE_NODE was always treated as the default\n\nI agree, though processing in order allows the operator more flexibility. I am not sure of an exact use case where they would want a processed-first-$COMPUTE_NODE file to take precedence, but there may be one...","accounts_in_message":[],"_revision_number":1},{"id":"d45b6a2078b0ae996a17cc057c6ced85f6900930","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-11-15 18:18:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e77d9ef1178241d756286ba630aac1cf07bf108c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 18:27:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12b4fe1068c4eae872cb4d62349714f : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2693bb5273a74bc6a8ed3caa12176a57 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5b1a3242dd3731360e664c2741442b68ffb3490c","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-11-21 00:06:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2475dcc1a38f854f42d8d9c46fdcfa18b4525abf","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-21 00:11:40.000000000","message":"Patch Set 3: Code-Review+2\n\nThis is good. We also talked about making it so runtime-detected conflicts would cause that provider to be ignored, with an error logged. But I\u0027m fine if that\u0027s left as an implementation detail.","accounts_in_message":[],"_revision_number":3},{"id":"1cdf0ea2c78cc3d3d20c2d05d8967bb8d431aa5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 00:15:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3c8f66cdbf47ccb69975db573fad36 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2950b0cb4024ee98c2bf41b2050586b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f3d165520d58b9bdf7680ad8d4668eebf8b8aaf7","author":{"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},"date":"2019-11-21 00:21:07.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9d91c19e4befb6e4784936980a6514dbbd574263","author":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"date":"2019-11-21 18:17:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dcbae80569c162d4505ac74598ed5e5b0028f12b","author":{"_account_id":29334,"name":"Dakshina Ilangovan","email":"dakshina.ilangovan@intel.com","username":"dakshinai"},"date":"2019-11-22 00:08:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b4f304176d36dfd4b6d5b3b79a2dc3357660aab","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-22 00:22:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbce087f9555ecb30e897092c8551ec9966e6951","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-01-13 20:39:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"093c8e5b0ae570390cf86e971144b72743225330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 20:39:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dd9498909560789f802081bc30d99de8718dbade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 20:55:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2928adb4314be0b93efc343829c056 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f71164f58b54788b678826d5e60111f : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b5aebf13b5d15a7dce4ac9dfdd6412de886a043e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 20:55:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"508cedbf37c5514427b362b2049351337e0808b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 20:57:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/cdd9bed3a74b4025b87dd6ec320dd03d : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7aaa79f019e6bd3c81f45712a4fab74cd127c217","revisions":{"b44811352ca34ad4b0ed622e742eb4f6e8a47726":{"kind":"REWORK","_number":1,"created":"2019-11-07 19:33:10.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/14/693414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/693414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/1"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-11-07 19:25:30.000000000","tz":-480},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-11-07 19:25:30.000000000","tz":-480},"subject":"Spec: Provider config YAML file","message":"Spec: Provider config YAML file\n\nDuring implementation it was found that there is no reliable method\nto identify identification conflicts between actual UUID/NAME and\nUUID:$COMPUTE_NODE before the point at which we cannot allow the loading\nprocess to fail. Since the spec previously required failure on an\nidentification conflict, I am proposing this change to ignore additional\nentries when an identification conflict is detected.\n\nChange-Id: If64509785675442ed9e8e8e452bf7dac837c1953\nBlueprint: provider-config-file\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b44811352ca34ad4b0ed622e742eb4f6e8a47726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b44811352ca34ad4b0ed622e742eb4f6e8a47726"}]},"branch":"refs/heads/master"},"abcce4bdd25997cc5e3baf12fe6f46bd970cbd9e":{"kind":"REWORK","_number":2,"created":"2019-11-15 18:18:47.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/14/693414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/693414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/2"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-11-07 19:25:30.000000000","tz":-480},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-11-15 18:14:09.000000000","tz":-480},"subject":"Update provider config spec for identification conflicts","message":"Update provider config spec for identification conflicts\n\nDuring implementation it was found that there is no reliable method\nto identify identification conflicts between actual UUID/NAME and\nUUID:$COMPUTE_NODE before the point at which we cannot allow the loading\nprocess to fail. Since the spec previously required failure on an\nidentification conflict, I am proposing this change to ignore additional\nentries when an identification conflict is detected.\n\nChange-Id: If64509785675442ed9e8e8e452bf7dac837c1953\nBlueprint: provider-config-file\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/abcce4bdd25997cc5e3baf12fe6f46bd970cbd9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/abcce4bdd25997cc5e3baf12fe6f46bd970cbd9e"}]},"branch":"refs/heads/master"},"7aaa79f019e6bd3c81f45712a4fab74cd127c217":{"kind":"REWORK","_number":3,"created":"2019-11-21 00:06:38.000000000","uploader":{"_account_id":29745,"name":"Dustin Cowles","email":"cowlesd@gmail.com","username":"dustinc","status":"inactive"},"ref":"refs/changes/14/693414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/693414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/693414/3"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-11-07 19:25:30.000000000","tz":-480},"committer":{"name":"Dustin Cowles","email":"dustin.cowles@intel.com","date":"2019-11-21 00:06:25.000000000","tz":-480},"subject":"Update provider config spec for identification conflicts","message":"Update provider config spec for identification conflicts\n\nDuring implementation it was found that there is no reliable method\nto identify identification conflicts between actual UUID/NAME and\nUUID:$COMPUTE_NODE before the point at which we cannot allow the loading\nprocess to fail. Since the spec previously required failure on an\nidentification conflict, I am proposing this change to allow for the\nuse of both by ignoring $COMPUTE_NODE entries if a UUID/NAME record\nalso exists.\n\nChange-Id: If64509785675442ed9e8e8e452bf7dac837c1953\nBlueprint: provider-config-file\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7aaa79f019e6bd3c81f45712a4fab74cd127c217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7aaa79f019e6bd3c81f45712a4fab74cd127c217"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
