)]}'
{"id":"openstack%2Fnova-specs~693844","triplet_id":"openstack%2Fnova-specs~master~I158080dab789de6acd06a8f73fd3bc6b72523a52","project":"openstack/nova-specs","branch":"master","topic":"bp/allow-secure-boot-for-qemu-kvm-guests","hashtags":[],"change_id":"I158080dab789de6acd06a8f73fd3bc6b72523a52","subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","status":"MERGED","created":"2019-11-12 13:37:21.000000000","updated":"2020-02-10 14:52:45.000000000","submitted":"2020-02-10 14:50:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":47,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693844-1581346246502-71032318","meta_rev_id":"b3fd700d46971599a5ba54892926e35ed514d5f6","_number":693844,"virtual_id_number":693844,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 14:50:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":2,"date":"2020-02-06 11:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2020-02-10 14:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-02-10 14:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 22:51:06.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-06 11:25:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-10 14:31:52.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-02-10 14:50:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b539fca9095cc282621fbd9c780e14b5717c6138","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-11-12 13:37:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3005bb8ad985cf03b4b92dc80d3077153a00c17f","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-11-12 13:38:27.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5eb24495ff5bca4d3b36ab142557029e3c7aee06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:44:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37d16d6499b4be1bc880d3546d1b14d : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96b917546ab4620ad109eb0372cf38a : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b3661e984b34cdb111093e80ec1ab7813bda92e7","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-01-15 17:27:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"05bff474c0af5fe01b813cca581b60f68d33bc6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 17:36:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0755bf888c4b3ea2f0653a58dba8e5 : FAILURE in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c45fcd2ea1b4f9c9fa5d021574e9401 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"1930ae657f5569d2307f0cca3d25ffaa5337dde5","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-01-16 10:20:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3e6a215f6752557bb03230e153614d65f0976153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 10:31:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ef6118d9cc40328a25b607c2e9b519 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f98dfdbe4c84c24bcaff54b65f7598a : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":4},{"id":"de8c974d8e2c15f209b473d07a43ee710b75d93b","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-01-16 12:53:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b0bffc9d5eb81f1e52718bef034297e90d928fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 13:04:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d244706061bf4805beb77205a04f4fe7 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ef5b155d09460ba23775a84b2c81a7 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":5},{"id":"ced22953fef9e0b94f229ee2bb26a1850f1cb974","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 11:47:24.000000000","message":"Patch Set 5: Code-Review-1\n\n(20 comments)\n\nProbably enough here to warrant the -1. Okay with the idea though","accounts_in_message":[],"_revision_number":5},{"id":"f46aba78aaa3fddca213470138ba5e831a4a360c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-27 13:41:37.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM. I will upgrade to +@ if Stephen\u0027s comments are fixed","accounts_in_message":[],"_revision_number":5},{"id":"5d182414cdfc363d2b18aa0b225b390170a36323","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-02-04 13:37:16.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"521b0758fa971ecf6dda97fb97344a10ae5c5ab8","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-02-04 14:51:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9f8f3566e439214e66f38b32fca000e76a610360","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-02-04 15:02:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fbe783501c90f7509cb2ec38999bdd97af879e2b","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-02-04 15:03:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6522484548efe459e359fa1524f67cb49841f5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:13:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446dea230c98472d95cdfbbf6350e669 : FAILURE in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2305733b0c924c5e86c92f4f7c4ca88a : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":6},{"id":"c52a454f15257d93338c83d032002bbdf3132214","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-02-04 16:18:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"22dc9e20b095f86ac27a252ae55bc2838e5a43a0","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-02-04 17:14:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8148393e877a0ca138ecd5a31c48be7b5f389eb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 17:27:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9f88c55e56455d8f261b3bb45cb651 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b91f7da291142f99d96f619c80ad8d1 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":7},{"id":"483272e6a63f1de5952708fe86fe900722262c41","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-06 11:08:33.000000000","message":"Patch Set 7: Code-Review-1\n\n(10 comments)\n\nCouple of nits but I think this is good to go now. Will quickly address those and we can move on","accounts_in_message":[],"_revision_number":7},{"id":"021338c53565a325fe71c7d496f9a747b863f3e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-06 11:24:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ee26b8be86c77dd628fbc5018debf6155b4b010d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-06 11:24:55.000000000","message":"Patch Set 7:\n\n(9 comments)\n\nAddressed my own nits","accounts_in_message":[],"_revision_number":7},{"id":"a9099d49644610e3c873c5eee20dc29e7021f175","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-06 11:25:00.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2b537ed865b6b2901c5ccd2461109ed9074df6a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 11:37:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/605caae4ffaf425f9876c6e483b950ff : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39d56c9480741e294f396edf05047c0 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":8},{"id":"09a9b292c03e13115ac7e80e985d62c4b58b7bb3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-10 14:31:52.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nThanks Kashyap!","accounts_in_message":[],"_revision_number":8},{"id":"4be038306ffbe43da30ea13e52041662195893b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:32:05.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"aa68897f5e8be03eba33c728c13f7f436d1b0408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:50:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"768c726e4d3224938b2f8a0d2e7c1df14f390224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:50:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5893bd29863419fb4e2658ad06fae55 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c27fe9f10d49818cff20b5e3c91042 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":8},{"id":"fdea0105209d3102de85c0bebcf3113eef6ec8d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:52:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/843d23fd6ae24b7f953d291710e4161d : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"007e08cbfb071aff030ca4bff93b7c98848c7de9","revisions":{"31da9f9c401bfac8c11a8f1a93d268ccd595ad00":{"kind":"REWORK","_number":1,"created":"2019-11-12 13:37:21.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/1"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:35:14.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" spec","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" spec\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/31da9f9c401bfac8c11a8f1a93d268ccd595ad00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/31da9f9c401bfac8c11a8f1a93d268ccd595ad00"}]},"branch":"refs/heads/master"},"4e04739666c18ca99ece7d92cd9616aeb2ab70f0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-12 13:38:27.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/2"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:37:46.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4e04739666c18ca99ece7d92cd9616aeb2ab70f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4e04739666c18ca99ece7d92cd9616aeb2ab70f0"}]},"branch":"refs/heads/master"},"d2bc9c643a0d51c7c97a78d43e7f8e41ab04a5ec":{"kind":"REWORK","_number":3,"created":"2020-01-15 17:27:53.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/3"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-15 17:23:11.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2bc9c643a0d51c7c97a78d43e7f8e41ab04a5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2bc9c643a0d51c7c97a78d43e7f8e41ab04a5ec"}]},"branch":"refs/heads/master"},"5ebac79b3d07edd6f92416b96714646879f82002":{"kind":"REWORK","_number":4,"created":"2020-01-16 10:20:52.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/4"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-16 10:20:08.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5ebac79b3d07edd6f92416b96714646879f82002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5ebac79b3d07edd6f92416b96714646879f82002"}]},"branch":"refs/heads/master"},"af8247a921b888a3a2fc0eb7ecc03dba3853264c":{"kind":"REWORK","_number":5,"created":"2020-01-16 12:53:24.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/5"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-16 12:36:07.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/af8247a921b888a3a2fc0eb7ecc03dba3853264c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/af8247a921b888a3a2fc0eb7ecc03dba3853264c"}]},"branch":"refs/heads/master"},"7d1263e7c8325fcbc3d027b67b59382ee6b1e0d6":{"kind":"REWORK","_number":6,"created":"2020-02-04 15:03:41.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/6"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-02-04 15:03:21.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d1263e7c8325fcbc3d027b67b59382ee6b1e0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d1263e7c8325fcbc3d027b67b59382ee6b1e0d6"}]},"branch":"refs/heads/master"},"7a6c553b9da870943784ced493543b6a3a5df4c6":{"kind":"REWORK","_number":7,"created":"2020-02-04 17:14:21.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/44/693844/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/7"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-02-04 17:13:44.000000000","tz":60},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7a6c553b9da870943784ced493543b6a3a5df4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7a6c553b9da870943784ced493543b6a3a5df4c6"}]},"branch":"refs/heads/master"},"007e08cbfb071aff030ca4bff93b7c98848c7de9":{"kind":"REWORK","_number":8,"created":"2020-02-06 11:24:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/693844/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/44/693844/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/44/693844/8"}}},"commit":{"parents":[{"commit":"20907c8141400d243d66c3653fe5fd348c987f3c","subject":"Merge \"Address comments on \u0027flavor-extra-spec-validators\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20907c8141400d243d66c3653fe5fd348c987f3c"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-11-12 13:32:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-02-06 11:24:21.000000000","tz":0},"subject":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri","message":"Re-propose \"Secure Boot support for KVM \u0026 QEMU guests\" for Ussuri\n\nPreviously-approved: Train\n\nBlueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I158080dab789de6acd06a8f73fd3bc6b72523a52\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/007e08cbfb071aff030ca4bff93b7c98848c7de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/007e08cbfb071aff030ca4bff93b7c98848c7de9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
