)]}'
{"id":"openstack%2Fnova-specs~699669","triplet_id":"openstack%2Fnova-specs~master~I52ea7a99856efee48aaf0b1f7a70705c7b24d850","project":"openstack/nova-specs","branch":"master","topic":"bp/action-event-fault-details","hashtags":[],"change_id":"I52ea7a99856efee48aaf0b1f7a70705c7b24d850","subject":"Add action event fault details","status":"MERGED","created":"2019-12-18 11:35:22.000000000","updated":"2020-03-25 14:11:27.000000000","submitted":"2020-01-30 14:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699669-1580393094220-f323dc7b","meta_rev_id":"7066cb92db024827f1cee4e9c0145f77ac450560","_number":699669,"virtual_id_number":699669,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 14:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2020-01-14 10:39:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-01-27 13:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-14 10:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-30 13:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 11:50:45.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-12-18 11:50:56.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-12-18 11:51:56.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-12-18 16:28:06.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2020-01-14 10:39:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-01-27 13:04:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 13:53:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-30 14:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67673b87022dc9a5e60bdb642aad4745230305fc","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-18 11:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7608aa3733ac445b29ba67466ee8bfbec15ee297","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-18 11:50:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c5dd937989d8bbe1982c8121f0e7b548887c04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 11:59:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44bad1e6a94e4f848b53cf917804e9b6 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b889275a1c4e2a919d04d06604a825 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f7ad5af4ba9d85bcc746835ff7deabb8bc21eb10","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-18 16:28:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"05adfba7ed04a7832485c45e97d1f994897f4aac","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-19 06:07:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e8bdb494ec70f8c8b17fb1aabfb0dcb353107fd","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-19 06:08:28.000000000","message":"Patch Set 2:\n\n(10 comments)\n\nThanks Matt.","accounts_in_message":[],"_revision_number":2},{"id":"66b73a5cdea51f4d67da86e54001f7eb76e56fd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 06:32:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5732aef10d7d45de8c38e73e2a274fdb : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a5b1e50796c44e99052ed56c6477b5e : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"18a3f38716a895ed35a9b664d13ed86a94ba55a1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-10 13:03:00.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nim kind of neutral on this.\n\ni am not sure is see a case where a non admin can do something\nactionable with the fault details to allow whatever action they were trying to succeed other then filing a ticket with the could admin.\n\nso im not sure this is a useful exetention to the api but i also see that it could give piece of mind to an end user because they might know why its went to error but equally frustrate them if they cant fix it.\n\nso as i said i dont really think this is useful for a non adming.\nfor an admin sure but non admin i dont think so. that said i dont think its particallrly harmful although i also wonder if this is somthing we would want a policy rule for so that an operator could decide if they want to expose it or not.","accounts_in_message":[],"_revision_number":2},{"id":"784f73a5abaac55c4675651f957236bb3721cb4a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-13 07:27:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bf1bc2efb0ff9414131bbf5fd2e5a27d8a633d0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-13 10:46:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nLet\u0027s add a policy so that the admin can decide if she exposes the extra fault information to the end user. Besides that I\u0027m OK with this.","accounts_in_message":[],"_revision_number":2},{"id":"4c548630c1f294ba934e066d1337fd7174e91890","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-13 14:39:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"509cd7e799922462c0600e61b4f57c2053ce3d20","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-14 08:06:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"087bd1c91c25a71cdb1465c88350ca462feb32b6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-14 08:47:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0c234b7b9a714be22ee91f3a74318839704c2aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 09:01:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7cdbab47864436caca15b318e329a4d : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7788abaaba496893e38b0cd0c7797a : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f604d90f599ddce29d4a57e9f03c05a4e10f63ad","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-14 10:25:00.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"af76e023be1106d41cc5684bed4bc20672041f79","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-14 10:39:23.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nim still not sure how useful this will be to end users but since we have a policy rule now for this im ok with this.","accounts_in_message":[],"_revision_number":3},{"id":"40ce5ab96fb2a429efff0aad4e9ca338c8b692a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 13:04:37.000000000","message":"Patch Set 3: Code-Review+2\n\nI\u0027m onboard with this","accounts_in_message":[],"_revision_number":3},{"id":"77446fa5c8dbeea2d67e60b5035935e2e7be0f11","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 13:53:00.000000000","message":"Patch Set 3: Workflow+1\n\nTwo core reviews agreed and two weeks passed since the last PS was proposed so I\u0027m going to approve it","accounts_in_message":[],"_revision_number":3},{"id":"7300d5e96c024128a470bbf2e3c9fdd481fb7dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:53:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ffb6a1a7bbaaa786e861774525dd4ec326b7a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:04:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3b3f8fe29f4bd7a6053d7f8fdc5765 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e618a8b502344a9855d35d048f6c4ef : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a7dff69ffc72cb4f88b19355acacd8a6b9f98632","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:04:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a55baabe08d738e8becf47adc5cc4647419c407d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:06:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/7399713b398a4d67acde840192fc575a : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3},{"id":"1036937fcfe530a9bb511049478ea9aa60e1b9fe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-25 02:19:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"127c83ceda5b0b4c89fb0930550781eed6c2a06e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-03-25 06:21:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bc93457c4013c9133a9faeb32ae94fb26d90f57b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-25 14:11:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"60fb46e3dff6ad1d0f47a748b7de9249919ab533","revisions":{"ada252e89bfa4bcb00962ec82315a5b0f604b680":{"kind":"REWORK","_number":1,"created":"2019-12-18 11:35:22.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/69/699669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/699669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/1"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-12-18 11:32:31.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-12-18 11:32:36.000000000","tz":480},"subject":"Add action event fault details","message":"Add action event fault details\n\nThe spec proposes to add the fault details to the failed instance\naction event.\n\nAPIImpact\n\nBlueprint: action-event-fault-details\n\nChange-Id: I52ea7a99856efee48aaf0b1f7a70705c7b24d850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ada252e89bfa4bcb00962ec82315a5b0f604b680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ada252e89bfa4bcb00962ec82315a5b0f604b680"}]},"branch":"refs/heads/master"},"73108e90498218d70f3a5a80842454a3a706a8d4":{"kind":"REWORK","_number":2,"created":"2019-12-19 06:07:29.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/69/699669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/699669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/2"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-12-18 11:32:31.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-12-19 06:07:26.000000000","tz":480},"subject":"Add action event fault details","message":"Add action event fault details\n\nThe spec proposes to add the fault details to the failed instance\naction event.\n\nAPIImpact\n\nBlueprint: action-event-fault-details\n\nChange-Id: I52ea7a99856efee48aaf0b1f7a70705c7b24d850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/73108e90498218d70f3a5a80842454a3a706a8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/73108e90498218d70f3a5a80842454a3a706a8d4"}]},"branch":"refs/heads/master"},"60fb46e3dff6ad1d0f47a748b7de9249919ab533":{"kind":"REWORK","_number":3,"created":"2020-01-14 08:47:25.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/69/699669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/699669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/699669/3"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-12-18 11:32:31.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-01-14 08:46:02.000000000","tz":0},"subject":"Add action event fault details","message":"Add action event fault details\n\nThe spec proposes to add the fault details to the failed instance\naction event.\n\nAPIImpact\n\nBlueprint: action-event-fault-details\n\nChange-Id: I52ea7a99856efee48aaf0b1f7a70705c7b24d850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/60fb46e3dff6ad1d0f47a748b7de9249919ab533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/60fb46e3dff6ad1d0f47a748b7de9249919ab533"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
