)]}'
{"id":"openstack%2Fnova-specs~701763","triplet_id":"openstack%2Fnova-specs~master~I288e4a2bd12702a1e7f7ebed544c95eb4a40e641","project":"openstack/nova-specs","branch":"master","topic":"bp/non-admin-filter-instance-by-az","hashtags":[],"change_id":"I288e4a2bd12702a1e7f7ebed544c95eb4a40e641","subject":"Non-Admin user can filter their instances by more filters","status":"MERGED","created":"2020-01-09 16:13:28.000000000","updated":"2020-01-30 13:08:26.000000000","submitted":"2020-01-30 13:02:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701763-1580389349631-ff1353d7","meta_rev_id":"16610b2e62de80df7e08659b734dde017b5ba1bf","_number":701763,"virtual_id_number":701763,"owner":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 13:02:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":2,"date":"2020-01-30 10:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-01-30 12:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-01-30 12:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 14:44:57.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-27 18:22:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-30 09:26:32.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-01-30 10:25:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 12:41:18.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-30 13:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"488b398545265c55dc0f02fa9b082ad08e2ca3e8","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-09 16:13:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6077aa156ea6119e438441d9cb68d89511f80409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 16:34:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e5d7bbe3e548f0a1d38a742e01c9f3 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a3524fdebb46368613f8403527952d : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a9ba16af06e64dd985c02b05570562e2275c596d","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-11 14:37:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3e8c0bd4a64967158e89f8caaf70f514e916439","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-11 14:38:12.000000000","message":"Patch Set 2:\n\nI added justification for the microversion bump and alternatives we discussed on IRC this week.","accounts_in_message":[],"_revision_number":2},{"id":"47b7e4e5143aa09e5e83af58dcb91c13d5332f13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 14:46:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0c13bcbbaa479eba0a1f7ea06ca0eb : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327321dd972f412189f76feabc96df46 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"af1a3c4a0b2e15d39b8de49d0191bae6cac1a5d1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-20 13:45:05.000000000","message":"Patch Set 2: Code-Review+1\n\nThis looks good to me. But needs a liaison. If nobody else is interested then you can add me as a liaison.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":2},{"id":"8fbdb9c440a5a574eb46a0386ff1a3294425c720","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-20 14:15:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nSoft -1, I think clear the non admin policy choices is make scense.\n@gibi, if nobody call you to the liaison, you can add me.","accounts_in_message":[],"_revision_number":2},{"id":"28a16ecc89f81b8591f9d48076be7b250ea8ca39","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-20 14:20:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81dba2f42a82c5c5852d709155a3e3cdf92aa234","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-20 14:42:08.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e (1 comment)\n\nAs the discuss with gmann, I think \"rule:project_reader_api\" role is ok.","accounts_in_message":[],"_revision_number":2},{"id":"550c28ab0d8e64796821d4a0599dbb7d0d7d7a51","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-20 14:45:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nI am -1 before we audit all filters and check if any other filter also required for non-admin.","accounts_in_message":[],"_revision_number":2},{"id":"ab37c2e6d31b2047e64a785842746f93adcfd1b7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-20 14:52:42.000000000","message":"Patch Set 2: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"28b0378207af7b3524f7d53d3b92693b78c7c94f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-20 15:07:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d17fc1d68af53c863683695bdee8e900756b3e8b","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-25 15:29:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"71395c5996dfd1fbed1689771fd47a548ccd519b","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-25 15:31:33.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"eff770deb20dcb23162aae84c4f39ee258c48d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 15:43:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18d977b214148a496cff72e601bab7b : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701a5affc90b43398a4c66d4d42df38b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e9421c26b0454fa40df23a812efe7c57bbe05ee5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 12:22:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)\n\nThis seems sane, but I do wonder why we can\u0027t just change the policy. This has probably been discussed already so it would be good to add that rationale in here, if so. The -1 is for that. The rest of the comments are style things that would be good to fix before merge","accounts_in_message":[],"_revision_number":4},{"id":"ec4f4de7c79fb2c4715e9387c44068be7ea25703","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-27 13:35:27.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"9855f12bda80ec8c8bbefd3c030b1356c00e94ed","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 17:26:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"062084880cc0a42265e16ce77478c3a2359a907a","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-27 17:47:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"339b457c1a160df2f462f1eee3601948dffd13c8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-27 18:22:03.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nfew more filter which can be discussed to add.","accounts_in_message":[],"_revision_number":4},{"id":"45688d93e2190e354c46489926fa0ef1325ac8f3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-28 09:07:37.000000000","message":"Patch Set 4: Code-Review+1\n\n@Thanks Gmann for the double check","accounts_in_message":[],"_revision_number":4},{"id":"09eb4b8bf363e7077686f280fca968ebb7592a2a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 11:48:37.000000000","message":"Patch Set 4:\n\nAs I understand it, this needs to be updated to reflect the additional filters that end-users can use and then we\u0027re good to go, yeah? @Victor are you able to do that so we can keep this moving?","accounts_in_message":[],"_revision_number":4},{"id":"b604c33c1d163936f5edde76b1f0a0e1e6284878","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-30 07:45:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"78832e10834158af624783c76e6340465000fe07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 07:53:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a9a6b68bc94ddd9827da905298c7ef : FAILURE in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec9a44c16b540f7b17e121673bb2941 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"3cbc3f1fafd77700af8addb386dcef5613c999b6","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-30 07:55:01.000000000","message":"Patch Set 4:\n\n(7 comments)\n\nThanks for all these reviews ! Is it what you expect or do you want me change something more ?","accounts_in_message":[],"_revision_number":4},{"id":"c2831f903226262964abf128b4da12984a59c58f","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-30 08:04:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"444d18d3526efa295f1d05c62b02df9670f18df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 08:20:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561d628eb37f45c2a1a5d2dfd587f62f : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e588f75933d4526997fef4a1fe52154 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":6},{"id":"fe7f95695b1430fb316310fe3ee00282bba0bb99","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 08:53:23.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"c5f89d8e9ac586535fb3bf2610e2c4e7a4e7347c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-30 09:26:32.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nLGTM over all, some nits inline.","accounts_in_message":[],"_revision_number":6},{"id":"44175bcb598dcaf9408263d2b8fb1c369e280a12","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-01-30 09:54:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b935c8d58ec1793709c83b47860e478beda0871c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:04:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aeea3f2bcfd4eb4906223b7d5481615 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6def69fd304cfb97550eada7daef3a : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":7},{"id":"d056e5b9b80379bf8a5dce6f4c6503281bb79ca0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 10:25:25.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\nThis looks good now. Thanks for the updates","accounts_in_message":[],"_revision_number":7},{"id":"9d88ee9099b517fe7e7eec91e2d5688fea9388fd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 12:41:18.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"81e464f31a3b2c81b5969206c002a2712aea97a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:41:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"852e34a10dda022b559e7a5d22b67ddf30282ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:02:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80cddc9839d74d07b3fff470b02ff8a1 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80598d5d60f249ff8c28369d67843855 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":7},{"id":"ab88bbae881a8ee5704f0916ae62a4c393a4d715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:02:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"8ca3bb96988af6d8d9c1cf4e99d0cd68a5d3f570","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:08:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/6a900cb4c59e4f09a758735d119e248e : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c76d1b3149de83cd8234642e258054d2e19d3ce3","revisions":{"7e666b16e066abada6b583013beed0bc77e6c3b1":{"kind":"REWORK","_number":1,"created":"2020-01-09 16:13:28.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/1"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:13:16.000000000","tz":60},"subject":"Non-admin user can filter their instances by AZs","message":"Non-admin user can filter their instances by AZs\n\nNova API specification does not allow user to filter their instances by\navailability zone, it is until now considered to a admin-only filter. As AZ\nare an input parameter at VM creation available to any user and it is an\nimportant notion for cloud usage, it should be easy for anyone to filter\nresources on it.\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e666b16e066abada6b583013beed0bc77e6c3b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e666b16e066abada6b583013beed0bc77e6c3b1"}]},"branch":"refs/heads/master"},"a6572a0334f24e7f1a203b5c8db18a3c5c9f7d40":{"kind":"REWORK","_number":2,"created":"2020-01-11 14:37:19.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/2"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-11 14:36:47.000000000","tz":60},"subject":"Non-admin user can filter their instances by AZs","message":"Non-admin user can filter their instances by AZs\n\nNova API specification does not allow user to filter their instances by\navailability zone, it is until now considered to a admin-only filter. As AZ\nare an input parameter at VM creation available to any user and it is an\nimportant notion for cloud usage, it should be easy for anyone to filter\nresources on it.\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a6572a0334f24e7f1a203b5c8db18a3c5c9f7d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a6572a0334f24e7f1a203b5c8db18a3c5c9f7d40"}]},"branch":"refs/heads/master"},"1815e848746e81397f41a52c65fd1bae3daffafd":{"kind":"REWORK","_number":3,"created":"2020-01-25 15:29:08.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/3"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-25 15:28:12.000000000","tz":60},"subject":"Non-admin user can filter their instances by AZs","message":"Non-admin user can filter their instances by AZs\n\nMany instances filter are restricted to admin-only users, while the related\nattribute are readable when showing instance detail for non admin users.\n\nIn order to stay coherent, all existing instance filters who are related to a\nfield readable by default to non admin users when showing instance details,\nshould be allowed by default without policy modification.\n\nAPIImpact\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1815e848746e81397f41a52c65fd1bae3daffafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1815e848746e81397f41a52c65fd1bae3daffafd"}]},"branch":"refs/heads/master"},"649afab26e0a14173e0760ee58775c60120477ab":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-25 15:31:33.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/4"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-25 15:29:44.000000000","tz":60},"subject":"Non-admin user can filter their instances by AZs","message":"Non-admin user can filter their instances by AZs\n\nMany instances filter are restricted to admin-only users, while the\nrelated attribute are readable when showing instance detail for non\nadmin users.\n\nIn order to stay coherent, all existing instance filters who are\nrelated to a field readable by default to non admin users when showing\ninstance details, should be allowed by default without policy\nmodification.\n\nAPIImpact\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/649afab26e0a14173e0760ee58775c60120477ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/649afab26e0a14173e0760ee58775c60120477ab"}]},"branch":"refs/heads/master"},"d68dc2883877fd48afecd562fc525022ef66579e":{"kind":"REWORK","_number":5,"created":"2020-01-30 07:45:46.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/5"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-30 07:45:32.000000000","tz":60},"subject":"Non-admin user can filter their instances by AZs","message":"Non-admin user can filter their instances by AZs\n\nMany instances filter are restricted to admin-only users, while the\nrelated attribute are readable when showing instance detail for non\nadmin users.\n\nIn order to stay coherent, all existing instance filters who are\nrelated to a field readable by default to non admin users when showing\ninstance details, should be allowed by default without policy\nmodification.\n\nAPIImpact\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d68dc2883877fd48afecd562fc525022ef66579e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d68dc2883877fd48afecd562fc525022ef66579e"}]},"branch":"refs/heads/master"},"033f9f01f403fc6f7c99a4de10f166693c1d558a":{"kind":"REWORK","_number":6,"created":"2020-01-30 08:04:22.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/6"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-30 08:04:06.000000000","tz":60},"subject":"Non-admin user can filter their instances by AZs","message":"Non-admin user can filter their instances by AZs\n\nMany instances filter are restricted to admin-only users, while the\nrelated attribute are readable when showing instance detail for non\nadmin users.\n\nIn order to stay coherent, all existing instance filters who are\nrelated to a field readable by default to non admin users when showing\ninstance details, should be allowed by default without policy\nmodification.\n\nAPIImpact\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/033f9f01f403fc6f7c99a4de10f166693c1d558a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/033f9f01f403fc6f7c99a4de10f166693c1d558a"}]},"branch":"refs/heads/master"},"c76d1b3149de83cd8234642e258054d2e19d3ce3":{"kind":"REWORK","_number":7,"created":"2020-01-30 09:54:50.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/63/701763/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/701763/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/701763/7"}}},"commit":{"parents":[{"commit":"e7769b0227106ecfd026936c57562e04400aff1b","subject":"Merge \"Amend cross-cell-resize spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7769b0227106ecfd026936c57562e04400aff1b"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-09 16:11:50.000000000","tz":60},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-01-30 09:53:44.000000000","tz":60},"subject":"Non-Admin user can filter their instances by more filters","message":"Non-Admin user can filter their instances by more filters\n\nMany instances filter are restricted to admin-only users, while the\nrelated attribute are readable when showing instance detail for non\nadmin users.\n\nIn order to stay coherent, all existing instance filters who are\nrelated to a field readable by default to non admin users when showing\ninstance details, should be allowed by default without policy\nmodification.\n\nAPIImpact\n\nChange-Id: I288e4a2bd12702a1e7f7ebed544c95eb4a40e641\nImplements: blueprint non-admin-filter-instance-by-az\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c76d1b3149de83cd8234642e258054d2e19d3ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c76d1b3149de83cd8234642e258054d2e19d3ce3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
