)]}'
{"id":"openstack%2Fnova-specs~702810","triplet_id":"openstack%2Fnova-specs~master~Ib9ad99527bb1de2aef52e236032b68a62de0d401","project":"openstack/nova-specs","branch":"master","topic":"bp/proposes","hashtags":[],"change_id":"Ib9ad99527bb1de2aef52e236032b68a62de0d401","subject":"Enable REST quiescing","status":"ABANDONED","created":"2020-01-16 07:51:12.000000000","updated":"2021-04-14 12:32:58.000000000","total_comment_count":63,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"27d12f35a9185c46e5d0897ae8ef9bf36b69be94","_number":702810,"virtual_id_number":702810,"owner":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-08-10 11:53:12.000000000","_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-23 06:51:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2020-08-06 11:02:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-08-03 21:37:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":-2,"date":"2020-08-06 17:43:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-23 06:36:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2020-07-23 06:36:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":-1,"date":"2020-07-23 06:36:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 12:54:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 16:44:06.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-07-08 17:20:51.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-07-08 19:28:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-08 20:50:26.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-09 09:14:57.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-07-09 15:25:54.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2020-07-23 06:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-03 21:37:55.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-08-06 11:02:02.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-08-06 17:43:09.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"e2767e2fe234aaf97ff5de105414b753a12c1df8","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-01-16 07:51:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f61c94c7bd8ebc80fe12b79bc4aa8a834f8c6a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:01:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfb90cc81ae4af88b6177363b10a578 : FAILURE in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8e08bad5fe48ec87c1df52b543849d : FAILURE in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a9745836c7590397676cddc018c79bf84b8a4eea","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-01-16 08:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb6289e854db32c90b19edaba9cbf565479d7aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:17:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4173a9e9874d1cbe855cdac1f161aa : FAILURE in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99bfb338a59435aa26c34eb934ac854 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"19a851eaab717c87908d669aebb967c83ed96738","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-01-16 08:26:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ba89bda33ba3998bb64d74f01c3a776d702356b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:35:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad76a3c83074319990e54a8475b0bce : FAILURE in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56828469b87c48deb48bae5f6a785ff4 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"7058c69ff2409c0c7ba79e74633191b645ff1280","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-01-16 09:14:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"126c57345a6ecb8fa22e62428b333345925462cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 09:29:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a806060dbe6c442ea784568f412fb606 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be250efbfd041c1a43542066bfe1097 : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"0122f281c968a45257b8c2bdc9876665a9a9c517","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-01-16 09:42:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8f2c0e11020023f3fcca478140ff2b6960b4fa1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 09:53:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d22ef591d61470d891f20325ab90a9f : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e016c34ba6244a9999b6b08563b9544 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":5},{"id":"03b4523a48f960696acb35fbc55f7e03cb335b45","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 12:54:53.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nThis seems sane from the libvirt side of things, but I\u0027d like someone else to look at it from the API side before I give my full ack","accounts_in_message":[],"_revision_number":5},{"id":"d60141c9351a7d9cd485dd2642a620d18f3e17aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 12:55:02.000000000","message":"Topic changed from bp/proposes to bp/on-demand-vm-quiescence","accounts_in_message":[],"_revision_number":5},{"id":"b196f4d25c4b8055412c790ebef9d76bd0cfd0c9","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-01-30 16:44:06.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a4a03268d9b824b285e724fa4591510c1b2ac884","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-01-30 16:46:33.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1838aeb062c5a4152fab0136bc5643de354ab1b6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 17:11:18.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\ncouple of things inline","accounts_in_message":[],"_revision_number":5},{"id":"4741e985683c623f72d8fea7d7e0549efc6bf3b5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 18:31:35.000000000","message":"Abandoned\n\nWe\u0027re at spec freeze. This has downvotes, and no response from the author since proposed.\n\nTo repropose for a subsequent $release please:\n- Restore (or ask a nova-core to do so)\n- Move the spec from specs/ussuri/ to specs/$release/\n- Update the blueprint\u0027s Series Goal","accounts_in_message":[],"_revision_number":5},{"id":"6242c345d56a99cec13c6d185e327abc1881eb78","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-06-11 07:24:37.000000000","message":"Patch Set 5:\n\n(11 comments)\n\nHi again, thanks for the comments/reviews. I\u0027ve deferred this blueprint to post-ussuri release on core advice.","accounts_in_message":[],"_revision_number":5},{"id":"58f7c94fc865447d26404dec5d3fbf633ae205bb","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-06-11 07:24:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"b58fae62fd1dc8d2909793f447e82cf036badbba","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-06-11 07:25:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fb662b68fb46271dda328270dd53cd6d3c29d25a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 07:32:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d3870e214f447da8e373b960a958ba : FAILURE in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047307e82e79431e941e2fece7499bd5 : SUCCESS in 5m 35s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/victoria/approved/on-demand-vm-quiescence.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/victoria/approved/on-demand-vm-quiescence.rst","accounts_in_message":[],"_revision_number":6},{"id":"b979e6dc2890e08d55c82a633c3b05ad628ab33c","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-06-11 08:18:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"53afb5270acd5de47f3482818828cba0c730eed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 08:30:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e54210e61f646e7ac1df14e1f6ac16f : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485d831eeb0e4cb199f4a07685fa2ebf : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":7},{"id":"abb0bef1b42c4f4d16249436fccffabefc3f0e64","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-11 12:32:14.000000000","message":"Patch Set 7: Code-Review-1\n\n(8 comments)\n\nI mostly replied to the PS5 comments to continue the discussion.","accounts_in_message":[],"_revision_number":7},{"id":"288e49e50e14aa1efef46473e6d4f50049a6042c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-06-13 15:36:33.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nDrive-by comment while I\u0027m out on paternity, I\u0027ll aim to review this more once I\u0027m back.","accounts_in_message":[],"_revision_number":7},{"id":"45b63d0cbc7133a09309ef52f268c934105796e2","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-06-18 08:59:57.000000000","message":"Patch Set 7: Workflow-1\n\nThanks for the comments. They sent me down a bit of a rabbit hole; I\u0027ll report back when I resurface.","accounts_in_message":[],"_revision_number":7},{"id":"eeab0a637e69dfae8cf43e960759f7e845850334","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-07-08 13:44:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"842d07f0a87378a6ad88538e1396e9992a5270f8","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-07-08 13:46:07.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c6c4895c3ebda595e8c39badafec2619f1064814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 14:05:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e0b72c87b8442fb3e13d039b9ffd6f : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deccebdbb8eb4371a39c866fe8998efb : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":8},{"id":"83698b8f5ca7cd524c28381ef16af9add2825af8","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-07-08 17:20:51.000000000","message":"Patch Set 8: Code-Review+1\n\nI think this is a good addition.  Fills a notable gap.","accounts_in_message":[],"_revision_number":8},{"id":"b7983e8bdde1243875e294e3b745fba3587fc163","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 19:28:10.000000000","message":"Patch Set 8:\n\nI like the idea and it has been requested quite a bit. I wonder about the usage though.\n\nIt sounds like the expectation is that cinder would be calling this api. I don\u0027t think that would be the case, nor what we want.\n\nBesides in general trying to avoid any hard dependencies in Cinder on Nova (we need to accommodate the use cases where nova is not involved) there is also the issue that there is not always a 1:1 mapping between cinder volumes and VMs. A VM may have a boot volume and one more data volumes. Some of these volumes may be from cinder (boot from volume case) and some may have a mix of ephemeral and Cinder volumes.\n\nIf all cinder volumes are in a volume group, then a snapshot can be requested on all of them together. But if they have not been grouped, I wonder what the expected behavior would be. Would we quiesce the entire VM, and hence all volumes, but only snapshot one? We then end up with the risk of inconsistent writes across the volumes and potential corruption. Granted, that\u0027s no worse than how it is today, but still a concern we should be aware of and think about how to handle.\n\nIf a backup application is using this, would it initiate the backup process through nova or through Cinder. I would think it would need to be through nova to be able to backup the entire VM. Other backup applications I have worked with operate on the VM, since that is the control point as far as being able to quiesce IO goes. It\u0027s also where we can know if IO can even be quiesced or not. Is the guest OS going to be involved in this process, or is the hypervisor going to be responsible for freezing the VM? If the guest is involved, OSs like Windows can actually reject a queiesce request using VSS if one of the writers is not able to hold off IO at that point. Or it can fail mid way through if it is not able to hold off on IO for the duration of the timeout window.\n\nWe may also need some APIs added to Cinder to support this, if it is coming from Nova to Cinder. Some backends need a \"prepare for snapshot\" operation to happen where it flushes local caches and gets ready to take a snapshot in order to complete the snapshot operation in time.\n\nOverall, I like the idea. But I think we need to flesh out some more details before we know if the proposal is going to work right in most cases.","accounts_in_message":[],"_revision_number":8},{"id":"847c6ecffdeb098638dbc32230fc85912fa782d1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-08 20:50:26.000000000","message":"Patch Set 8:\n\nA few observations (with the caveat that I don\u0027t really work on Nova).\n\nFirst, I agree with Sean\u0027s comments about the likely order this would be used (Nova first, Cinder later).\n\nI wonder whether concentrating on the Cinder use case is misleading for this spec, particularly when you mention the possibility of an \"external backup toool\" using the API to quiesce VMs.\n\nThe use case discussed in #previousattempt is to be able to make application consistent snapshots of a VNF spanning multiple VMs.  If you want external tools to use this, I don\u0027t know that it should use the os-server-external-events API, which by my understanding is supposed to be used only by other openstack services.  (The policy config can allow other actors to use it, so maybe that\u0027s OK.  And the way it works, i.e., to issue a bunch of events to multiple VMs does fit this use case.)  But the main issue I see is that your proposal isn\u0027t fine-grained enough for this scenario: my backup tool issues quiesce requests to VMs 1, 2, and 3; before actually doing the backup, it may need to verify that each VM is actually quiesced (not just that the quiesce request was accepted).  Maybe this operation is so fast that you don\u0027t need to worry about this, but I wonder whether you really want the task states to be \u0027quiescing\u0027 (preparing, but not ready yet), \u0027quiesced\u0027 (ready for snapshotting), \u0027unquiescing\u0027 (returning to normal).  I guess what it comes down to is whether nova will return \"completed\" or \"in-progress\" in response to these requests.\n\nThe other thing I\u0027m wondering about is the auto-cleanup of unquiesced instances.  Being crass, why would an operator care if the instances are sitting around quiesced as long as the customer is paying for them?  Surely an external tool would check to make sure everything is back up?  Also, the VNF case as described in #previousattempt mentioned that re-enabling the component VMs sometimes needs to happen in a particular order.  So just turning instances back on in the order they show up in the database may do more harm than good.\n\nBut overall, I think this is functionality that users will like.","accounts_in_message":[],"_revision_number":8},{"id":"f070d6aa4fc9bd0c36c09723ce784dc3ed04c5c1","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-07-09 09:14:57.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1322cf93255c8150272ebcbf950f2f2e3561ac8f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-09 12:28:33.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9701fbcc30aee2a977a70f5f80f1c3df431b646f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-09 15:24:12.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1efb8c7a677e671a6c079f244b0912b65441767e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-09 15:25:54.000000000","message":"Patch Set 8: Code-Review-1\n\nSo I think I agree with Brian, I actually like the previous spec we approved. Could you please address in more detail why you thing that appoach isn\u0027t a good idea.\n\nWe have things similar to auto-confirm-resize for the unquiese problem. I would hope we follow a similar pattern.","accounts_in_message":[],"_revision_number":8},{"id":"69e4a82475be928e2477f6e921e55d08f70af166","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-14 09:59:20.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"7a0c86aea421c168b2c6fbe9d24c2c3ca57d019b","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-07-14 13:21:22.000000000","message":"Patch Set 8:\n\n(7 comments)\n\n\u003e I like the idea and it has been requested quite a bit. I wonder\n \u003e about the usage though.\n \u003e \n \u003e It sounds like the expectation is that cinder would be calling this\n \u003e api. I don\u0027t think that would be the case, nor what we want.\n \u003e \n \u003e Besides in general trying to avoid any hard dependencies in Cinder\n \u003e on Nova (we need to accommodate the use cases where nova is not\n \u003e involved) there is also the issue that there is not always a 1:1\n \u003e mapping between cinder volumes and VMs. A VM may have a boot volume\n \u003e and one more data volumes. Some of these volumes may be from cinder\n \u003e (boot from volume case) and some may have a mix of ephemeral and\n \u003e Cinder volumes.\n \u003e \n \u003e If all cinder volumes are in a volume group, then a snapshot can be\n \u003e requested on all of them together. But if they have not been\n \u003e grouped, I wonder what the expected behavior would be. Would we\n \u003e quiesce the entire VM, and hence all volumes, but only snapshot\n \u003e one? We then end up with the risk of inconsistent writes across the\n \u003e volumes and potential corruption. Granted, that\u0027s no worse than how\n \u003e it is today, but still a concern we should be aware of and think\n \u003e about how to handle.\n\n Only entire VMs can be quiesced as far as I know and I don\u0027t really see\n a reason why it should (or could) be otherwise. You want consistent writes\n across all volumes in a VM and you can never be sure what\u0027s going on in the \n FS when you freeze it, which means you can never know which specific volumes\n need to be frozen (whatever that may mean in a given context).\n\n \u003e If a backup application is using this, would it initiate the backup\n \u003e process through nova or through Cinder. I would think it would need\n \u003e to be through nova to be able to backup the entire VM. Other backup\n \u003e applications I have worked with operate on the VM, since that is\n \u003e the control point as far as being able to quiesce IO goes. It\u0027s\n \u003e also where we can know if IO can even be quiesced or not. Is the\n \u003e guest OS going to be involved in this process, or is the hypervisor\n \u003e going to be responsible for freezing the VM? If the guest is\n \u003e involved, OSs like Windows can actually reject a queiesce request\n \u003e using VSS if one of the writers is not able to hold off IO at that\n \u003e point. Or it can fail mid way through if it is not able to hold off\n \u003e on IO for the duration of the timeout window.\n \n The idea is that it doesn\u0027t need to use either. It simply uses Nova\n to quiesce an instance, do it\u0027s thing and unquiesce it again. Any failure\n of the quiescing operation needs to be handled by the invoking application.\n\n \u003e We may also need some APIs added to Cinder to support this, if it\n \u003e is coming from Nova to Cinder. Some backends need a \"prepare for\n \u003e snapshot\" operation to happen where it flushes local caches and\n \u003e gets ready to take a snapshot in order to complete the snapshot\n \u003e operation in time.\n\n Again, the backup process itself doesn\u0027t necessarily start from within openstack\n nor is it the subject of this spec.\n  \n \u003e Overall, I like the idea. But I think we need to flesh out some\n \u003e more details before we know if the proposal is going to work right\n \u003e in most cases.\n\n\n \u003e A few observations (with the caveat that I don\u0027t really work on\n \u003e Nova).\n \u003e \n \u003e First, I agree with Sean\u0027s comments about the likely order this\n \u003e would be used (Nova first, Cinder later).\n \u003e \n \u003e I wonder whether concentrating on the Cinder use case is misleading\n \u003e for this spec, particularly when you mention the possibility of an\n \u003e \"external backup toool\" using the API to quiesce VMs.\n\n It would appear so. There seems to be this idea of the proposed changes being\n intimately tied to Cinder, which isn\u0027t the case. The aim is to give API \n consumers the ability to get an instance into a state of quiescence - no more,\n no less - for whatever reason, not just backup creation.\n \n I originally only mentioned Cinder as the motivating use case because that\u0027s\n what I specifically need it for. The spec itself was drafted as tool agnostic.\n\n \u003e The use case discussed in #previousattempt is to be able to make\n \u003e application consistent snapshots of a VNF spanning multiple VMs. \n \u003e If you want external tools to use this, I don\u0027t know that it should\n \u003e use the os-server-external-events API, which by my understanding is\n \u003e supposed to be used only by other openstack services.  (The policy\n \u003e config can allow other actors to use it, so maybe that\u0027s OK.  And\n \u003e the way it works, i.e., to issue a bunch of events to multiple VMs\n \u003e does fit this use case.)  But the main issue I see is that your\n \u003e proposal isn\u0027t fine-grained enough for this scenario: my backup\n \u003e tool issues quiesce requests to VMs 1, 2, and 3; before actually\n \u003e doing the backup, it may need to verify that each VM is actually\n \u003e quiesced (not just that the quiesce request was accepted).  Maybe\n \u003e this operation is so fast that you don\u0027t need to worry about this,\n \u003e but I wonder whether you really want the task states to be\n \u003e \u0027quiescing\u0027 (preparing, but not ready yet), \u0027quiesced\u0027 (ready for\n \u003e snapshotting), \u0027unquiescing\u0027 (returning to normal).  I guess what\n \u003e it comes down to is whether nova will return \"completed\" or\n \u003e \"in-progress\" in response to these requests.\n \n Yeah, this has to do with the async-vs-sync nature of the operation. I\u0027ve been\n going back and forth on this and ultimately came down on the side of the \n quiescence operation being fast enough to be considered synchronous. The VSS on\n Windows even puts a 10 second limit on quiescence duration. Obviously, the \n biggest upside to this is simpler implementation: fewer changes to the state \n machine, fewer API endpoints etc. \n \n All that said, I\u0027m not super married to the idea. If the overwhelming opinion\n is that we treat the operation as asynchronous with all that entails, I don\u0027t\n really have a problem with that; I\u0027ve outlined a series of alternatives which\n could be utilized in that case.\n\n \u003e The other thing I\u0027m wondering about is the auto-cleanup of\n \u003e unquiesced instances.  Being crass, why would an operator care if\n \u003e the instances are sitting around quiesced as long as the customer\n \u003e is paying for them?  Surely an external tool would check to make\n \u003e sure everything is back up?  Also, the VNF case as described in\n \u003e #previousattempt mentioned that re-enabling the component VMs\n \u003e sometimes needs to happen in a particular order.  So just turning\n \u003e instances back on in the order they show up in the database may do\n \u003e more harm than good.\n\n The idea behind the autocleanup mechanism is that an instance stuck in a state\n of quiescence for too long could do a whole lot of (catastrophic) damage and\n most cases would benefit from a \"better safe than sorry\" approach. For everyone\n else we can provide a way of turning it off.","accounts_in_message":[],"_revision_number":8},{"id":"7eac6eaa0af2e6ddb145e1cc40a332af31cd0ee2","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-07-23 06:36:41.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"247e46ec1a0ed0bdf80b6fe19bea25f66f91b076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 06:51:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255a51622256451b9bf8e6294cbb3d41 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ab5ecb8e84c4935a7c668753db01ba6 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":9},{"id":"6b938588f6e40ee9a650c9e060a4ca36df8fe9e7","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-08-03 21:37:55.000000000","message":"Patch Set 9: Code-Review+1\n\nThanks Marek for clearing up some of my confusion, I\u0027m still not fully convinced that this can be a sync API given os-server-external-events is casting to the computes but I\u0027m not going to let that block things.\n\nAs highlighted on the ML we are now past spec freeze but please feel free to write up a request to have this merged as an exception. I\u0027m +1 for now until that is done.\n\nPlease also feel free to post code for review ahead of that being accepted.","accounts_in_message":[],"_revision_number":9},{"id":"030eb76a3a34291c121341b73ba568216289e88c","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-08-05 06:38:12.000000000","message":"Patch Set 9:\n\nAlright, thank you for your help. I\u0027ll take a look at writing that exception request (any tips would be appreciated) and I\u0027ll get to work on implementation ASAP.\n\nAs for the sync-vs-async thing, I completely understand where you\u0027re coming from. I think we should be able to switch midstream to the full blown async. implementation (as discussed above) easily enough  if we find that the simplified sync version simply isn\u0027t cutting it.","accounts_in_message":[],"_revision_number":9},{"id":"e0f88f99f4a61ef08980579259d95087338a5b31","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-06 11:02:02.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a944eeb132c9f36b563a717a86b52728d0e38d48","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-06 11:02:31.000000000","message":"Patch Set 9:\n\nother then my comments inline im ok with the idea if its a server action.","accounts_in_message":[],"_revision_number":9},{"id":"b1c59d1c62518c6a1477720e6874dba9dee3ac3f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-06 14:56:13.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\nReading the comments makes me wondering about the about the use case.","accounts_in_message":[],"_revision_number":9},{"id":"e7541fb9925f6c825c1dcb6c5b5a51907f570589","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-06 17:43:09.000000000","message":"Patch Set 9: Code-Review-2\n\nProcedural -2 due to spec freeze.\n\nAs we discussed on the today nova meeting[1] there is an open question to be sorted out about the use case. If there is a use case where quiescing is called outside of OpenStack then that should be listed in the spec and then os-external-events are not a good API. But if the only valid use case is to snapshot Cinder volumes, os-external-events are a good API.\n\nAs we had the spec deadline last Thurday and we still have an open question about the use case I decided not to grant the freeze exception. \n\nPlease continue discussing this spec and we can approve it for early W cycle.\n\n[1] http://eavesdrop.openstack.org/meetings/nova/2020/nova.2020-08-06-16.00.log.html#l-105","accounts_in_message":[],"_revision_number":9},{"id":"01eae960f306c6ded9c945addc43191805258259","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-08-10 11:53:12.000000000","message":"Patch Set 9:\n\n\u003e (1 comment)\n \u003e \n \u003e Reading the comments makes me wondering about the about the use\n \u003e case.\n\nAlright, fair enough. Thanks for taking a look, I\u0027ll fix up the proposal and move it to W.","accounts_in_message":[],"_revision_number":9},{"id":"a97a31914a1cd58d1c9a2e262082ffc3c40b71a5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-01 14:35:43.000000000","message":"Patch Set 9:\n\nping: is this spec still pushed for wallaby?","accounts_in_message":[],"_revision_number":9},{"id":"27d12f35a9185c46e5d0897ae8ef9bf36b69be94","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 12:32:58.000000000","message":"Abandoned\n\nIt seems it is not pushed any more so I abandon it. Feel free to restore it or ping me (gibi) on IRC to do so.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"d8bdb6d7fead53e8b250285bc56b4892e547e035","revisions":{"85d50fb3c222967389e3aab0efe1da82ca21d802":{"kind":"REWORK","_number":1,"created":"2020-01-16 07:51:12.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/1"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-06 15:10:09.000000000","tz":60},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-16 07:51:03.000000000","tz":60},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/85d50fb3c222967389e3aab0efe1da82ca21d802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/85d50fb3c222967389e3aab0efe1da82ca21d802"}]},"branch":"refs/heads/master"},"857cbf7d8cd9554079992cea66393fd7a8b6edc6":{"kind":"REWORK","_number":2,"created":"2020-01-16 08:07:16.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/2"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-06 15:10:09.000000000","tz":60},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-16 08:06:48.000000000","tz":60},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/857cbf7d8cd9554079992cea66393fd7a8b6edc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/857cbf7d8cd9554079992cea66393fd7a8b6edc6"}]},"branch":"refs/heads/master"},"b6c7972f76f67ecbf3485d5129d83ffe411d47b5":{"kind":"REWORK","_number":3,"created":"2020-01-16 08:26:28.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/3"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-06 15:10:09.000000000","tz":60},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-16 08:26:14.000000000","tz":60},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b6c7972f76f67ecbf3485d5129d83ffe411d47b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b6c7972f76f67ecbf3485d5129d83ffe411d47b5"}]},"branch":"refs/heads/master"},"2c8d8443f9ab78e1f2794c920d78578d5402c171":{"kind":"REWORK","_number":4,"created":"2020-01-16 09:14:47.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/4"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-06 15:10:09.000000000","tz":60},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-16 09:14:31.000000000","tz":60},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2c8d8443f9ab78e1f2794c920d78578d5402c171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2c8d8443f9ab78e1f2794c920d78578d5402c171"}]},"branch":"refs/heads/master"},"5be495335922dc67b52067af0be03000f9e8d30c":{"kind":"REWORK","_number":5,"created":"2020-01-16 09:42:50.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/5"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-06 15:10:09.000000000","tz":60},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-01-16 09:42:35.000000000","tz":60},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5be495335922dc67b52067af0be03000f9e8d30c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5be495335922dc67b52067af0be03000f9e8d30c"}]},"branch":"refs/heads/master"},"86ab8fbca0e58f5ae4dc1d8ef0907a19d85f9bae":{"kind":"REWORK","_number":6,"created":"2020-06-11 07:25:02.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/6"}}},"commit":{"parents":[{"commit":"b259a414ae6f12175f20de5318e2df1da5de20c7","subject":"Merge \"Encrypted Emulated Virtual TPM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b259a414ae6f12175f20de5318e2df1da5de20c7"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-06-11 07:18:07.000000000","tz":120},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-06-11 07:18:07.000000000","tz":120},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/86ab8fbca0e58f5ae4dc1d8ef0907a19d85f9bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/86ab8fbca0e58f5ae4dc1d8ef0907a19d85f9bae"}]},"branch":"refs/heads/master"},"972210bd55809c9e9b0a5e872df81a3f2766b519":{"kind":"REWORK","_number":7,"created":"2020-06-11 08:18:36.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/7"}}},"commit":{"parents":[{"commit":"b259a414ae6f12175f20de5318e2df1da5de20c7","subject":"Merge \"Encrypted Emulated Virtual TPM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b259a414ae6f12175f20de5318e2df1da5de20c7"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-06-11 07:18:07.000000000","tz":120},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-06-11 08:18:20.000000000","tz":120},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/972210bd55809c9e9b0a5e872df81a3f2766b519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/972210bd55809c9e9b0a5e872df81a3f2766b519"}]},"branch":"refs/heads/master"},"bb31ecd6951f1ba5543222476c0f8de160c3fb25":{"kind":"REWORK","_number":8,"created":"2020-07-08 13:44:47.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/8"}}},"commit":{"parents":[{"commit":"b259a414ae6f12175f20de5318e2df1da5de20c7","subject":"Merge \"Encrypted Emulated Virtual TPM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b259a414ae6f12175f20de5318e2df1da5de20c7"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-06-11 07:18:07.000000000","tz":120},"committer":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-07-08 13:44:24.000000000","tz":120},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bb31ecd6951f1ba5543222476c0f8de160c3fb25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bb31ecd6951f1ba5543222476c0f8de160c3fb25"}]},"branch":"refs/heads/master"},"d8bdb6d7fead53e8b250285bc56b4892e547e035":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-07-23 06:36:41.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/10/702810/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/10/702810/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/10/702810/9"}}},"commit":{"parents":[{"commit":"7ddca43229f1e5bca01389e2bca1e4a595242ec3","subject":"Merge \"Support SRIOV interface attach and detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ddca43229f1e5bca01389e2bca1e4a595242ec3"}]}],"author":{"name":"mareklycka","email":"marek.lycka@ultimum.io","date":"2020-06-11 07:18:07.000000000","tz":120},"committer":{"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","date":"2020-07-23 06:36:41.000000000","tz":0},"subject":"Enable REST quiescing","message":"Enable REST quiescing\n\nThis blueprint proposes on demand instance quiescing using REST APIs.\n\nChange-Id: Ib9ad99527bb1de2aef52e236032b68a62de0d401\nBlueprint: on-demand-vm-quiescence\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d8bdb6d7fead53e8b250285bc56b4892e547e035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d8bdb6d7fead53e8b250285bc56b4892e547e035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
