)]}'
{"id":"openstack%2Fnova-specs~702943","triplet_id":"openstack%2Fnova-specs~master~Ib125c9de60ac614b4b8cb3ad0f03a4141efd0d2a","project":"openstack/nova-specs","branch":"master","topic":"bp/vgpu-multiple-types","hashtags":[],"change_id":"Ib125c9de60ac614b4b8cb3ad0f03a4141efd0d2a","subject":"Re-proposes multiple vGPU types in libvirt","status":"MERGED","created":"2020-01-16 18:01:59.000000000","updated":"2020-01-29 09:30:37.000000000","submitted":"2020-01-27 14:11:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702943-1580134304444-bbef5f2b","meta_rev_id":"356e339483e9926cfeac5f826d5ab46ac2b5e6f1","_number":702943,"virtual_id_number":702943,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 14:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-01-27 11:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-01-27 13:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-01-27 13:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 11:51:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-27 13:58:29.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-27 14:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5abb0402fc456a016b39c8ba91cecc5bddcbdd18","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-16 18:01:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4318f23af1fc921a4b57f4c3797466c07990697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:17:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07902fc03d446dfb0157e16f8188df1 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7596075b3390403cae507986f9e29469 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"387022b6318fb4ba2b636fe6607260c186f8c91d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 09:45:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)\n\nThis looks pretty good to me and is mostly a reproposal fwict. Just a couple of small questions inline along with some nits (the -1 is for the latter)","accounts_in_message":[],"_revision_number":1},{"id":"df5b4c0998d3494be017fd50bfe02f87d448fb80","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-27 10:38:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f73e05a318e26c88bb946272b46e19fb185b477b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 10:50:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4079dbe3c6274b419e89734eae475490 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56d6bb851074968b075f78540ed2ae9 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a8a02c7ee87f7e467c2dac67db631f24b000927a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-27 11:51:17.000000000","message":"Patch Set 2: Code-Review+2\n\nGiven the \"staticness\" and limited configurability available for these vGPUs at the moment, this makes senses. We will probably need to reassess things in the future if/when GPUs with vGPU support gain the ability to configure multiple GPUs at once, but since that hardware doesn\u0027t exist yet it\u0027s not worth worrying about","accounts_in_message":[],"_revision_number":2},{"id":"d696625650ae9ded1d9e213077592b9e10329566","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-27 13:58:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43d788c4355f9b1ab0adb5a57cb48059b625cd4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:58:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5c731157e9b8d66db00262c716d6cd85ae18177e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:11:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679a8e619c6a42c788a130f66cb56bec : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f84f74af7640b1ac8ad5b9ca019781 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e06ce4e2040efa29dfc5dc366ac2eb3d55595bf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:11:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cc52723ff14a12f93cdb21c86a7e065b35d92b87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:15:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/8f0d1bb97a414517848d5c560b5af1c8 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"cd9d85f9fc8038c54a4dccbdd64f0e84b8042590","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-29 09:30:37.000000000","message":"Patch Set 1:\n\n(9 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":2,"current_revision":"dee2ff3afd344cbab9689b95e79324f3bd69f152","revisions":{"7cae268c6b4c4531004ab3b53d8b161be100314a":{"kind":"REWORK","_number":1,"created":"2020-01-16 18:01:59.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/43/702943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/702943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/1"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-01-16 17:37:04.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-01-16 17:42:27.000000000","tz":60},"subject":"Re-proposes multiple vGPU types in libvirt","message":"Re-proposes multiple vGPU types in libvirt\n\nThis spec is for telling libvirt how to manage multiple vGPU types.\nThe only modification to the approved spec in Stein is to mention\nanother alternative and no longer mention upgrades since GPUs are\nalready modeled as children RPs.\n\nChange-Id: Ib125c9de60ac614b4b8cb3ad0f03a4141efd0d2a\nPartially-Implements: bp/vgpu-multiple-types\nPreviously-Approved: Stein\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7cae268c6b4c4531004ab3b53d8b161be100314a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7cae268c6b4c4531004ab3b53d8b161be100314a"}]},"branch":"refs/heads/master"},"dee2ff3afd344cbab9689b95e79324f3bd69f152":{"kind":"REWORK","_number":2,"created":"2020-01-27 10:38:48.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/43/702943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/702943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/702943/2"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-01-16 17:37:04.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-01-27 10:32:32.000000000","tz":60},"subject":"Re-proposes multiple vGPU types in libvirt","message":"Re-proposes multiple vGPU types in libvirt\n\nThis spec is for telling libvirt how to manage multiple vGPU types.\nThe only modification to the approved spec in Stein is to mention\nanother alternative and no longer mention upgrades since GPUs are\nalready modeled as children RPs.\n\nChange-Id: Ib125c9de60ac614b4b8cb3ad0f03a4141efd0d2a\nPartially-Implements: bp/vgpu-multiple-types\nPreviously-Approved: Stein\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dee2ff3afd344cbab9689b95e79324f3bd69f152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dee2ff3afd344cbab9689b95e79324f3bd69f152"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
