)]}'
{"id":"openstack%2Fnova-specs~722194","triplet_id":"openstack%2Fnova-specs~master~I319c2b7e90874820fb98736701398b9466d754c3","project":"openstack/nova-specs","branch":"master","topic":"bp/destroy-instance-with-datavolume","hashtags":[],"change_id":"I319c2b7e90874820fb98736701398b9466d754c3","subject":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination","status":"MERGED","created":"2020-04-23 02:58:37.000000000","updated":"2020-04-25 23:49:22.000000000","submitted":"2020-04-24 16:36:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722194-1587746213179-602cea05","meta_rev_id":"7fd9e57f030bcfcf703c93b4df1090b5015ebba7","_number":722194,"virtual_id_number":722194,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 16:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2020-04-24 14:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2020-04-24 16:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2020-04-24 16:14:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 02:59:20.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-04-24 14:35:25.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-04-24 16:14:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-24 16:36:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8ac4a598fef6cd3cf6d40e30454b50e70dada84","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-23 02:58:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8473caa7b63a36f20e306a134f07c86c9b7f0887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 03:11:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330820d843ad4affb065a450d5737efa : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ca972b4fac4b128e1565519314a962 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2253314c9490d9fb557988a44897da0fdfe0046f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-23 04:08:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b4e80decaa9cab46af49e0e3ac76c7756c7f245c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 04:17:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d17bdacc5f4b929800b03844e1edd6 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e54a885896c4171bfc88d8e46639a22 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8bc4525841a199a5983ee1e55be39bad0cecae1b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-04-23 13:56:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1b7fb384515ceb33836f2c5425111462fcaf0675","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-24 01:45:25.000000000","message":"Patch Set 2:\n\n(12 comments)\n\nThanks Dan, addressed all you comments, why am I so careless this time... need to think twice.","accounts_in_message":[],"_revision_number":2},{"id":"5134d1870387f04dd94454aee6f9aea121061d51","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-24 01:46:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d6edda3e1db4f114d4a026ef45f846823ca805a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 01:58:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16a62e9b0444c2a930f1ae2f89c5810 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1fc8ff674b641968e44ceab65418906 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"becd231f35980405ddb0ff9ffeb956f29fc3ad5f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-04-24 14:33:10.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b4fb6d816092439bf088217af950774eebfd9f12","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-04-24 14:34:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1e7b370dc0d53527c0667dd83dcdeabeecf9ee93","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-04-24 14:35:25.000000000","message":"Patch Set 4: Code-Review+2\n\nJust made a few grammar nit changes.","accounts_in_message":[],"_revision_number":4},{"id":"566f5ed319b24d36764d5da5a31a504ab10028ce","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-24 14:43:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ac8de01a2ee554bd89926babf1aa88735acc81e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:48:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8458a47fd81c42518ed8e678d2a968c7 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87d51f215ca4c68acfcc4882bbeb600 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"87e20e4812f5bf4e4121ca1d9719e55325e6878f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-24 16:14:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)\n\nthanks for update.","accounts_in_message":[],"_revision_number":4},{"id":"3b46ba96b4aa9f648aed480ed1de80b784f47780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:19:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"48ed0fbdc711dd502c87cc03334cfc5b411afca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:36:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564629da2a76457cbbb6f20500721de4 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f9c6cafc2e9480785f7d88f02de5da8 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"1a1709170e83b75a546ebcd251eb8df8d8933b0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:36:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0355b08fbfed6a6d866eda4f54624cc1b05071d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:41:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e47d313fa5b04f719bc642e51d0aae1a : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"56fb7ee89a815edbca8d6f68f1e0d27c9651d4dc","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-25 23:49:22.000000000","message":"Patch Set 4:\n\nThanks dansmith, gmann.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2084981c641de8442561ea1d95f1437e24c123a3","revisions":{"1d8f0ce5a3639d9e9c015b3b02534368ead8d7ed":{"kind":"REWORK","_number":1,"created":"2020-04-23 02:58:37.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/94/722194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/94/722194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/1"}}},"commit":{"parents":[{"commit":"93dc9ec0d507480c751910e3f9126c3277070ae6","subject":"Merge \"Adds spec infrastructure for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93dc9ec0d507480c751910e3f9126c3277070ae6"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-23 02:51:34.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-23 02:51:38.000000000","tz":480},"subject":"Update bp/destroy-instance-with-datavolume SPEC","message":"Update bp/destroy-instance-with-datavolume SPEC\n\nAs talked in [1], instead of using PATCH to update attached volume, use\nPUT API (update a volume attachmet API) instead.\n\n[1]https://review.opendev.org/#/c/693828/19/nova/api/openstack/wsgi.py@54\n\nBlueprint: destroy-instance-with-datavolume\n\nChange-Id: I319c2b7e90874820fb98736701398b9466d754c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8f0ce5a3639d9e9c015b3b02534368ead8d7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8f0ce5a3639d9e9c015b3b02534368ead8d7ed"}]},"branch":"refs/heads/master"},"ce33cc9c925d24f083a613f181df40de467cd338":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-23 04:08:01.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/94/722194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/94/722194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/2"}}},"commit":{"parents":[{"commit":"93dc9ec0d507480c751910e3f9126c3277070ae6","subject":"Merge \"Adds spec infrastructure for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93dc9ec0d507480c751910e3f9126c3277070ae6"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-23 02:51:34.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-23 04:05:32.000000000","tz":480},"subject":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination","message":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\n\nAs talked in [1], using PUT (update a volume attachmet API) instead of using\nPATCH to update attached volume.\n\n[1]https://review.opendev.org/#/c/693828/19/nova/api/openstack/wsgi.py@54\n\nBlueprint: destroy-instance-with-datavolume\n\nChange-Id: I319c2b7e90874820fb98736701398b9466d754c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ce33cc9c925d24f083a613f181df40de467cd338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ce33cc9c925d24f083a613f181df40de467cd338"}]},"branch":"refs/heads/master"},"3f83e2be2c8d1683332a49d7e59b58ab6d50b584":{"kind":"REWORK","_number":3,"created":"2020-04-24 01:46:12.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/94/722194/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/94/722194/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/3"}}},"commit":{"parents":[{"commit":"93dc9ec0d507480c751910e3f9126c3277070ae6","subject":"Merge \"Adds spec infrastructure for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93dc9ec0d507480c751910e3f9126c3277070ae6"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-23 02:51:34.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-24 01:45:46.000000000","tz":480},"subject":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination","message":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\n\nAs talked in [1], using PUT (update a volume attachmet API) instead of using\nPATCH to update attached volume.\n\n[1]https://review.opendev.org/#/c/693828/19/nova/api/openstack/wsgi.py@54\n\nBlueprint: destroy-instance-with-datavolume\n\nChange-Id: I319c2b7e90874820fb98736701398b9466d754c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3f83e2be2c8d1683332a49d7e59b58ab6d50b584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3f83e2be2c8d1683332a49d7e59b58ab6d50b584"}]},"branch":"refs/heads/master"},"2084981c641de8442561ea1d95f1437e24c123a3":{"kind":"REWORK","_number":4,"created":"2020-04-24 14:34:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/94/722194/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/94/722194/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/94/722194/4"}}},"commit":{"parents":[{"commit":"93dc9ec0d507480c751910e3f9126c3277070ae6","subject":"Merge \"Adds spec infrastructure for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93dc9ec0d507480c751910e3f9126c3277070ae6"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-23 02:51:34.000000000","tz":480},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2020-04-24 14:34:51.000000000","tz":0},"subject":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination","message":"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\n\nAs talked in [1], using PUT (update a volume attachmet API) instead of using\nPATCH to update attached volume.\n\n[1]https://review.opendev.org/#/c/693828/19/nova/api/openstack/wsgi.py@54\n\nBlueprint: destroy-instance-with-datavolume\n\nChange-Id: I319c2b7e90874820fb98736701398b9466d754c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2084981c641de8442561ea1d95f1437e24c123a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2084981c641de8442561ea1d95f1437e24c123a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
