)]}'
{"id":"openstack%2Fnova-specs~725801","triplet_id":"openstack%2Fnova-specs~master~Ic75c41fc1d19d5a8bd74d820624935e3ccff703f","project":"openstack/nova-specs","branch":"master","topic":"bp/use-pcpu-and-vcpu-in-one-instance","hashtags":[],"change_id":"Ic75c41fc1d19d5a8bd74d820624935e3ccff703f","subject":"Use PCPU and VCPU in one instance","status":"MERGED","created":"2020-05-06 09:12:20.000000000","updated":"2020-05-18 12:44:31.000000000","submitted":"2020-05-18 12:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725801-1589805749607-d1623aff","meta_rev_id":"d4f73b5997a203a9822d87f6c9041b3bf42c4f5c","_number":725801,"virtual_id_number":725801,"owner":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 12:42:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":2,"date":"2020-05-07 12:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-07 12:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":1,"date":"2020-05-18 12:33:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 12:39:44.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-05-18 12:33:46.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-18 12:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2db411ceb61f202ffc37c15fb1fc17e8d327454b","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-05-06 09:12:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d9b15ed6de3c4fe3d88d70742b9b45759461298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:25:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09c4ba87da34499baa843572d40773b : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0729fab215404c8ba18421b64605b810 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3297881c87cf7194fe6eaae069011a2b6bca29f9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-06 10:08:54.000000000","message":"Patch Set 1: Code-Review+2\n\nI reviewed this extensively for Ussuri and this still LGTM","accounts_in_message":[],"_revision_number":1},{"id":"dc8f47486190f3e31a33cc68670f5f78abb54929","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-05-07 07:11:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dfc0e855d0e88e10b66e03afb825117ff111813f","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-05-07 07:30:05.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHi Alex,\n\nThanks for your review. Please see my answers in the document.\n\nBr\nHuaqiang","accounts_in_message":[],"_revision_number":1},{"id":"8719d9446bd35259b33ffe951f6228646c4024c0","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-05-07 07:39:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ae191d5094ae01e551e8443ecc853674cf6d93b","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-05-07 09:34:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"483ed586321fac5c46393e7e1341b851ae0c854f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 09:44:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0f04b0790f4a3eb900c9681f5c6732 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939d086c395e4642ae8c574402b3b493 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1d92c929613292336461d75dfdaa152976ff36eb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-05-07 12:39:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bc95c9a23e49e96d391d6163d398ff094f974744","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-07 12:41:45.000000000","message":"Patch Set 2: Code-Review+2\n\nWill hold off +Wing because I\u0027m not actually sure if we can approve stuff rn","accounts_in_message":[],"_revision_number":2},{"id":"391763df9848ff5cdbd41fa5751018dafe970b3b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-18 12:33:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f88efea9c1115d1c8341701850900374434f92d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 12:33:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f7fa25ff3becd1050dc64b3790b9e0597cd6a19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 12:42:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ee8039e4f1c8deb3f924a3e4a8242b13b8795145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 12:42:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938d26c49acc4dbba1b520649eb33545 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cff41c9b1f4c859e422ad695fb25f8 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"1d795ac06d2d7830214d8906ade004ac3f738021","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 12:44:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e893f6e2805c4a6e8b2c04c3dfcc178c : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c588c8f05cfdae984d5ac44ce513c01b37f3b3df","revisions":{"93368221e8a93d6a6b5f0a352dd7d7481d08c4e0":{"kind":"REWORK","_number":1,"created":"2020-05-06 09:12:20.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/01/725801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/725801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/1"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-05-06 08:47:58.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-05-06 09:11:57.000000000","tz":480},"subject":"Use PCPU and VCPU in one instance","message":"Use PCPU and VCPU in one instance\n\nWe\u0027d like to propose a way to create instance to\ntake using of PCPU and VCPU resource in one instance.\n\nRe-proposed in victoria.\n\nChange-Id: Ic75c41fc1d19d5a8bd74d820624935e3ccff703f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93368221e8a93d6a6b5f0a352dd7d7481d08c4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93368221e8a93d6a6b5f0a352dd7d7481d08c4e0"}]},"branch":"refs/heads/master"},"c588c8f05cfdae984d5ac44ce513c01b37f3b3df":{"kind":"REWORK","_number":2,"created":"2020-05-07 09:34:57.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/01/725801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/725801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/725801/2"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-05-06 08:47:58.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-05-07 09:32:08.000000000","tz":480},"subject":"Use PCPU and VCPU in one instance","message":"Use PCPU and VCPU in one instance\n\nWe\u0027d like to propose a way to create instance to\ntake using of PCPU and VCPU resource in one instance.\n\nRe-proposed in victoria.\n\nChange-Id: Ic75c41fc1d19d5a8bd74d820624935e3ccff703f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c588c8f05cfdae984d5ac44ce513c01b37f3b3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c588c8f05cfdae984d5ac44ce513c01b37f3b3df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
