)]}'
{"id":"openstack%2Fnova-specs~728009","triplet_id":"openstack%2Fnova-specs~master~I93d78fff517c49d76059b2dea6acac5e2a86f40c","project":"openstack/nova-specs","branch":"master","topic":"bp/numa-topology-with-rps","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-06 11:21:07.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I93d78fff517c49d76059b2dea6acac5e2a86f40c","subject":"re-propose numa with placement","status":"MERGED","created":"2020-05-14 10:09:06.000000000","updated":"2025-01-06 11:21:07.000000000","submitted":"2020-05-22 09:24:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"728009-1590139484982-7f2e0e33","meta_rev_id":"0a2aa5a78da77bdde635b2f9e5c45a594ff569b6","_number":728009,"virtual_id_number":728009,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 09:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-05-22 09:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-20 07:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-14 12:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-05-22 09:13:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 12:26:25.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-05-20 07:22:03.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-05-22 09:13:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-22 09:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-17 18:15:07.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"}],"messages":[{"id":"97073a30fb3a9181476b6e77f3d9b89b546f8872","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-14 10:09:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9557d24eca6a32ee7380afe83037dd40dee1e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 10:23:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd37ef241d243399003c17dfce72b28 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d045bf336874846811c9069316de76e : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7e2c02cd26eff7e3d62bf1c8c5afafa88445812e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-14 10:24:37.000000000","message":"Topic changed from bp/numa-topolofy-with-rps to bp/numa-topoloy-with-rps","accounts_in_message":[],"_revision_number":1},{"id":"266122a91b4f23ff5e07cb3d8b728db4b2fc76c7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-05-14 12:26:25.000000000","message":"Patch Set 1: Code-Review+2\n\nSean just flipped the contributor names between me and him since he\u0027ll be working on this one for Victoria mainly.\n\n[sbauza@sbauza nova-specs]$ diff  specs/ussuri/approved/numa-topology-with-rps.rst specs/victoria/approved/numa-topology-with-rps.rst \n595d594\n\u003c * bauzas\n596a596,597\n\u003e * bauzas\n\u003e \n638a640,652\n\u003e \n\u003e History\n\u003e \u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\u003e \n\u003e .. list-table:: Revisions\n\u003e    :header-rows: 1\n\u003e \n\u003e    * - Release Name\n\u003e      - Description\n\u003e    * - Ussuri\n\u003e      - Introduced\n\u003e    * - Victoria\n\u003e      - Re-proposed","accounts_in_message":[],"_revision_number":1},{"id":"77ee5d86a62a60a29ad7b2fb5dad8c82f71dc257","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-05-14 12:27:39.000000000","message":"Topic changed from bp/numa-topoloy-with-rps to bp/numa-topology-with-rps","accounts_in_message":[],"_revision_number":1},{"id":"77f9bff54d9c9876a6d72fd2765aa59a6746943e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-05-20 07:22:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"441717065d97e4be4ede0a8236404871b156fd06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-22 09:13:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db5cd55b36ab86470c1c6b8363f550bf131b5630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:13:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e060f30d2fe53956793aff67a1131dee57e6c2b5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-22 09:13:39.000000000","message":"Patch Set 1:\n\nHopefully we actually get something done on this this time :-D","accounts_in_message":[],"_revision_number":1},{"id":"a1d6096167b60ab207b2e3ff076892b1109a186d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:24:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61719989b52460dba33beb075bfa526 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022c0750b4e74886991ff2ac26d45dee : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0700e73efb1cdbf36152dba795f07ba10f8997d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:24:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d217d0431200fdc4aab75fb3fdba2cdfaaa1f1de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:29:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/6d3cfa283d514baebb34b7794a63deac : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d4ad26148d9cc54363aa45422dc66d1e3783107f","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-12-17 18:15:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a2aa5a78da77bdde635b2f9e5c45a594ff569b6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-06 11:21:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e290c1fcacea614fd0be5ad2dd8f25b19f5487bd","revisions":{"e290c1fcacea614fd0be5ad2dd8f25b19f5487bd":{"kind":"REWORK","_number":1,"created":"2020-05-14 10:09:06.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/09/728009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/09/728009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/09/728009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/09/728009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/09/728009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/09/728009/1"}}},"commit":{"parents":[{"commit":"3bc3142246bb49a3acff7d71ecf89a881c049a01","subject":"Merge \"Re-proposes \u0027Proposal for a safer remote console with password authentication\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3bc3142246bb49a3acff7d71ecf89a881c049a01"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-05-14 10:07:12.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-05-14 10:09:04.000000000","tz":0},"subject":"re-propose numa with placement","message":"re-propose numa with placement\n\nThis patch updates the histroy and re-proposes\nthe spec for the victoria cycle.\n\nChange-Id: I93d78fff517c49d76059b2dea6acac5e2a86f40c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e290c1fcacea614fd0be5ad2dd8f25b19f5487bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e290c1fcacea614fd0be5ad2dd8f25b19f5487bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
