)]}'
{"id":"openstack%2Fnova-specs~728778","triplet_id":"openstack%2Fnova-specs~master~I65de5f8310a52cf6c8c9b2eec3085c17bef39a45","project":"openstack/nova-specs","branch":"master","hashtags":[],"change_id":"I65de5f8310a52cf6c8c9b2eec3085c17bef39a45","subject":"Support volume local cache","status":"MERGED","created":"2020-05-18 08:28:50.000000000","updated":"2020-06-29 09:11:50.000000000","submitted":"2020-06-29 09:09:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728778-1593421745691-ccd9e89b","meta_rev_id":"f8bf63a93eccfc72c35b3a60bc4005023fdeb7f9","_number":728778,"virtual_id_number":728778,"owner":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 09:09:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2020-05-26 16:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-29 08:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-29 08:59:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 16:09:44.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-06-29 08:59:29.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-06-29 09:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"643ff2a6d9d0dddea2ba6f1a2d95c82ca9172682","author":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"date":"2020-05-18 08:28:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c7d9d64ffabe4612fbdc44be8859d5b5ea86170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 08:39:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60967903f6254a9fb95610aa691d4ea8 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85e5102124c42e5a728225170a4e8ae : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c367f0f5fa4705709f6984f291b76f2313cc1390","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-26 16:09:44.000000000","message":"Patch Set 1: Code-Review+2\n\nSimple reproprose","accounts_in_message":[],"_revision_number":1},{"id":"54d828a5310d9fe0cd1f5176714883b1db0df746","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-06-29 08:59:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9bf609694dbda540c7728121cb07029117be82c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:59:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12a94753ae63ddb34a754dd31ad59baae4dd0afa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:09:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"672e3451b284a7cd9629dd185be77acd67807c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:09:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae23dc3f411f4311b64805550f2e0cc8 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d02311e9fe44ffa5db0720c4ae4a44 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f8bf63a93eccfc72c35b3a60bc4005023fdeb7f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:11:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/dd8ac646cb864bebb8ffc3a0ccf68378 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ad4ec3e9ceb9ca4c07df797b9a02ae3b4b5913c","revisions":{"9ad4ec3e9ceb9ca4c07df797b9a02ae3b4b5913c":{"kind":"REWORK","_number":1,"created":"2020-05-18 08:28:50.000000000","uploader":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"ref":"refs/changes/78/728778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/78/728778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/728778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/728778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/78/728778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/78/728778/1"}}},"commit":{"parents":[{"commit":"3bc3142246bb49a3acff7d71ecf89a881c049a01","subject":"Merge \"Re-proposes \u0027Proposal for a safer remote console with password authentication\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3bc3142246bb49a3acff7d71ecf89a881c049a01"}]}],"author":{"name":"Liang Fang","email":"liang.a.fang@intel.com","date":"2020-05-18 08:14:16.000000000","tz":480},"committer":{"name":"Liang Fang","email":"liang.a.fang@intel.com","date":"2020-05-18 08:14:16.000000000","tz":480},"subject":"Support volume local cache","message":"Support volume local cache\n\nMoved the spec from U release\n\nChange-Id: I65de5f8310a52cf6c8c9b2eec3085c17bef39a45\nSigned-off-by: Liang Fang \u003cliang.a.fang@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9ad4ec3e9ceb9ca4c07df797b9a02ae3b4b5913c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9ad4ec3e9ceb9ca4c07df797b9a02ae3b4b5913c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
