)]}'
{"id":"openstack%2Fnova-specs~730638","triplet_id":"openstack%2Fnova-specs~master~I3769eb16fe396404113eba75a64d86e6b3d86f81","project":"openstack/nova-specs","branch":"master","topic":"victoria-template","hashtags":[],"change_id":"I3769eb16fe396404113eba75a64d86e6b3d86f81","subject":"Make Feature Liaison optional","status":"MERGED","created":"2020-05-25 15:37:42.000000000","updated":"2020-10-21 02:24:29.000000000","submitted":"2020-06-04 12:42:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730638-1591274533608-91d87e12","meta_rev_id":"e627d6c34581d6fb7e5aaa8de3cf910e3be736e9","_number":730638,"virtual_id_number":730638,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 12:42:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-06-04 11:48:29.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-06-04 11:49:35.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 12:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2020-06-04 12:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2020-10-21 02:24:29.000000000","post_submit":true,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2020-06-04 12:32:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 12:28:51.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-05-29 17:46:58.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2020-06-01 09:46:18.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2020-06-04 11:49:35.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-06-04 12:22:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-06-04 12:32:09.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-06-04 12:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-21 02:24:29.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"6a610dc1db13660ebbbac8cca32d37bbef599e76","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-25 15:37:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c51b5c6d087212a4fe984aaf41c2427d4245dd3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-25 15:38:22.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for more input here on the ML","accounts_in_message":[],"_revision_number":1},{"id":"77c3268daabd063e79282381eee2c9c43f79833f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 15:50:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d0cd419b9049199c3de927afc83914 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/947274a6315d4877a6655b4f88fb06f2 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ec0652f2b404d537babc3da677886cc7d7acca75","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-25 18:40:43.000000000","message":"Patch Set 1:\n\nYup, this is good. Let’s get the main readme fixed up as well: https://specs.openstack.org/openstack/nova-specs/readme.html#feature-liaison-faq","accounts_in_message":[],"_revision_number":1},{"id":"e6df1b65d69b521260b3f3d25104d4b1da6ef184","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-26 08:09:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61c94ec92959e29f54c035d70ce7c2761b462ef5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-26 08:10:12.000000000","message":"Patch Set 2:\n\n\u003e Yup, this is good. Let’s get the main readme fixed up as well:\n \u003e https://specs.openstack.org/openstack/nova-specs/readme.html#feature-liaison-faq\n\nMost of that write up is still valid. I did add a note now that we made it optional in Victoria","accounts_in_message":[],"_revision_number":2},{"id":"75bac5571fb14346c2d423db2fb06d6500b1d113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:20:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbfc46ff6e84c9c9a0e91fa10859f20 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fa5d64f25a47cf85e90d0953ded365 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b6e4cb70662f0df1e690f499415f79abdeb259e0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-05-26 08:21:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e94fda5eab5b49ecc39e4ca59dfa8c9f487c8927","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-26 08:29:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"61749e7624c296ca5a2e3cf9567036cf57111761","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-26 08:29:56.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f5428de19ebc10764e3adcd635e852c153eaeea5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-05-26 08:37:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5e62577f2a884ee6c748272949eed52a40a32fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:50:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96049d822bb3467da60bfb805891875a : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0446a8a2470e40cbb90827b1384f4736 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d8581cebaff77d84a7a21460b71cfe657e5e302e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-26 12:28:51.000000000","message":"Patch Set 4: Code-Review+1\n\nyep this seams to be inline with the mailing list conversation so this looks good to me","accounts_in_message":[],"_revision_number":4},{"id":"1233559df9be85a241809590f76061d2bbf804a3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-05-26 14:33:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f9c987a278ba3b5fb8aff739c6570996a11a4d35","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-26 19:19:02.000000000","message":"Patch Set 4: Code-Review+2\n\nlgtm and as per what being discussed on ML thread.","accounts_in_message":[],"_revision_number":4},{"id":"00218eb0a882b8cf4b82882c5a7d3ff5e26c3f8f","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-05-29 17:46:58.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d5a0566972be0a235c1aacd8d93c013cbe8c61a4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-06-01 09:46:18.000000000","message":"Patch Set 4: Code-Review+2\n\nSeems like the right call, given how things worked out.","accounts_in_message":[],"_revision_number":4},{"id":"7d0eac9f2234a9103f16fded9fd2b0c0c4cf4384","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-03 16:10:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a094a3d6266e009deb66cd7e502e7eb299a6d58e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-04 11:48:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a7963cb0820f9d2d2d39de13d4590c59cdd9e877","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-04 11:48:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"196bbcceab81c00ec9b10d301c558a4a52955eae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-06-04 11:49:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d5075fb17013efd50aa1911b13b86afdd815536a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:59:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068d6fd3c7f6466e83e19af00b76c7a0 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2c3166108d741a98a6af4fcac7bb23b : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":5},{"id":"22f1f9d3f73d05169f4137311baa5d43dc69d19c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-04 12:22:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"72cd9add2492d9ecdc7a5be6f5e88fe4030f0998","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-06-04 12:32:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLet\u0027s merge it","accounts_in_message":[],"_revision_number":5},{"id":"c2a1a2870f200edf4366375abdede3389893e95b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:32:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b72885585a174c098d0f903d29180dbb7e29ccb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:42:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"44e9cc555a0d5a174f32f83d99b3953c9b65abe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:42:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23425347359478785c2d281ca9769bd : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1789459cdb7e4676aa1b67bf56253005 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":5},{"id":"f7706c1a23193b113904371b1986e85b1c549da2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:45:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/c00b52835fb14cf4b5965818eee68b2c : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1032743e5410088b770bb454eda794b6f1d50a99","revisions":{"3505954feb700f7a7980281791741faf13f5838e":{"kind":"REWORK","_number":1,"created":"2020-05-25 15:37:42.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/38/730638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/38/730638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/1"}}},"commit":{"parents":[{"commit":"ad3192d320079138124527002f8203d736fbb0bd","subject":"Merge \"re-propose numa with placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad3192d320079138124527002f8203d736fbb0bd"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-25 15:34:42.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-25 15:34:42.000000000","tz":120},"subject":"Make Feature Liaison optional","message":"Make Feature Liaison optional\n\nAs we discussed in the pre-PTG ML thread we don\u0027t see too much value\nkeeping the Feature Liaison mandatory in the spec but we keep it as an\noptional way for new contributors to declare that they would like to be\nmentored.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014917.html\n\nChange-Id: I3769eb16fe396404113eba75a64d86e6b3d86f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3505954feb700f7a7980281791741faf13f5838e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3505954feb700f7a7980281791741faf13f5838e"}]},"branch":"refs/heads/master"},"b8771b3f0b28fd67f931cd4ca5a43cf5a5a2cd81":{"kind":"REWORK","_number":2,"created":"2020-05-26 08:09:20.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/38/730638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/38/730638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/2"}}},"commit":{"parents":[{"commit":"ad3192d320079138124527002f8203d736fbb0bd","subject":"Merge \"re-propose numa with placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad3192d320079138124527002f8203d736fbb0bd"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-25 15:34:42.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-26 08:09:00.000000000","tz":120},"subject":"Make Feature Liaison optional","message":"Make Feature Liaison optional\n\nAs we discussed in the pre-PTG ML thread we don\u0027t see too much value\nkeeping the Feature Liaison mandatory in the spec but we keep it as an\noptional way for new contributors to declare that they would like to be\nmentored.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014917.html\n\nChange-Id: I3769eb16fe396404113eba75a64d86e6b3d86f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b8771b3f0b28fd67f931cd4ca5a43cf5a5a2cd81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b8771b3f0b28fd67f931cd4ca5a43cf5a5a2cd81"}]},"branch":"refs/heads/master"},"2a221ac7f1527c284125ea44496af06c368c89cf":{"kind":"REWORK","_number":3,"created":"2020-05-26 08:29:37.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/38/730638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/38/730638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/3"}}},"commit":{"parents":[{"commit":"ad3192d320079138124527002f8203d736fbb0bd","subject":"Merge \"re-propose numa with placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad3192d320079138124527002f8203d736fbb0bd"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-25 15:34:42.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-26 08:29:27.000000000","tz":0},"subject":"Make Feature Liaison optional","message":"Make Feature Liaison optional\n\nAs we discussed in the pre-PTG ML thread we don\u0027t see too much value\nkeeping the Feature Liaison mandatory in the spec but we keep it as an\noptional way for new contributors to declare that they would like to be\nmentored.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014917.html\n\nChange-Id: I3769eb16fe396404113eba75a64d86e6b3d86f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2a221ac7f1527c284125ea44496af06c368c89cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2a221ac7f1527c284125ea44496af06c368c89cf"}]},"branch":"refs/heads/master"},"16f2f937a229761aa736fd760a11ea0aae9abd57":{"kind":"REWORK","_number":4,"created":"2020-05-26 08:37:37.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/38/730638/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/38/730638/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/4"}}},"commit":{"parents":[{"commit":"ad3192d320079138124527002f8203d736fbb0bd","subject":"Merge \"re-propose numa with placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad3192d320079138124527002f8203d736fbb0bd"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-25 15:34:42.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-26 08:37:34.000000000","tz":0},"subject":"Make Feature Liaison optional","message":"Make Feature Liaison optional\n\nAs we discussed in the pre-PTG ML thread we don\u0027t see too much value\nkeeping the Feature Liaison mandatory in the spec but we keep it as an\noptional way for new contributors to declare that they would like to be\nmentored.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014917.html\n\nChange-Id: I3769eb16fe396404113eba75a64d86e6b3d86f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/16f2f937a229761aa736fd760a11ea0aae9abd57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/16f2f937a229761aa736fd760a11ea0aae9abd57"}]},"branch":"refs/heads/master"},"1032743e5410088b770bb454eda794b6f1d50a99":{"kind":"REWORK","_number":5,"created":"2020-06-04 11:48:08.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/38/730638/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/38/730638/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/38/730638/5"}}},"commit":{"parents":[{"commit":"ad3192d320079138124527002f8203d736fbb0bd","subject":"Merge \"re-propose numa with placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ad3192d320079138124527002f8203d736fbb0bd"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-05-25 15:34:42.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-06-04 11:46:44.000000000","tz":120},"subject":"Make Feature Liaison optional","message":"Make Feature Liaison optional\n\nAs we discussed in the pre-PTG ML thread and on the PTG[2] we don\u0027t see too\nmuch value keeping the Feature Liaison mandatory in the spec but we keep\nit as an optional way for new contributors to declare that they would like\nto be mentored.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014917.html\n[2] https://etherpad.opendev.org/p/nova-victoria-ptg\n\nChange-Id: I3769eb16fe396404113eba75a64d86e6b3d86f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1032743e5410088b770bb454eda794b6f1d50a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1032743e5410088b770bb454eda794b6f1d50a99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
