)]}'
{"id":"openstack%2Fnova-specs~736422","triplet_id":"openstack%2Fnova-specs~master~I96528ae95a8914431850039c93c4a24b303f73ef","project":"openstack/nova-specs","branch":"master","topic":"nova-support-multiple-boot-volume-with-boot-order-selection","hashtags":[],"change_id":"I96528ae95a8914431850039c93c4a24b303f73ef","subject":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec fix document error","status":"ABANDONED","created":"2020-06-18 03:37:12.000000000","updated":"2020-06-25 03:42:56.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98e5fe4baaf34ab6d87831cbc7b14282d90e9aad","_number":736422,"virtual_id_number":736422,"owner":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-18 05:52:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":-1,"date":"2020-06-25 03:14:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 05:52:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-25 03:14:58.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"4aa1a9ea009f532aa9fc3d392a7bea4836739592","author":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"date":"2020-06-18 03:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa91d123541502764a1d37c4116d5fb464643d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 03:42:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0dc56d3c34455aa70a8035aec04b39 : FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93b8679b9cf4cf08a5730c54ac77e0f : FAILURE in 4m 09s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/rocky/approved/nova-support-multiple-boot-volume-with-boot-order-selection.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/rocky/approved/nova-support-multiple-boot-volume-with-boot-order-selection.rst","accounts_in_message":[],"_revision_number":1},{"id":"eed8bbff6c9e1f85bd7d35ac8082529febffc6d1","author":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"date":"2020-06-18 05:07:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"073ee36cabeb103dbf3ebeaf2343b5009736d3a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 05:15:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488d810020a1467d9b3988e0fe261bbd : FAILURE in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a991c4591748487c99e39a5617f5e727 : SUCCESS in 4m 31s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/rocky/approved/nova-support-multiple-boot-volume-with-boot-order-selection.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/rocky/approved/nova-support-multiple-boot-volume-with-boot-order-selection.rst","accounts_in_message":[],"_revision_number":2},{"id":"9921b93fbace2d64d3187e18de135239c29fabe9","author":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"date":"2020-06-18 05:43:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29832a94ecff7cfc69bc8d0deb4f193927ccb0c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 05:52:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0250b41b5d114114894fd7af16d89646 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53cff5ae63b7406abc2f61d316fa4e39 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"5ba2c20a32c8b0be42483a5720bd42841c673860","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-06-25 03:14:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nIt\u0027s not valid to propose a spec for an older version of nova. Only specs for the Victoria release are being accepted. You will need to update this spec to target the victoria specs directory.","accounts_in_message":[],"_revision_number":3},{"id":"e4e713e7f441d7e92d0a74b1f7f7a83b87936263","author":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"date":"2020-06-25 03:42:56.000000000","message":"Abandoned\n\nUpdate to target the victoria specs directory","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"811e4b42559ec3cfb17b36d8f79b7c7fc16dff99","revisions":{"6900909a45b880ebec9ae9d72f43268b7a9e369a":{"kind":"REWORK","_number":1,"created":"2020-06-18 03:37:12.000000000","uploader":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"ref":"refs/changes/22/736422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/22/736422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/1"}}},"commit":{"parents":[{"commit":"b259a414ae6f12175f20de5318e2df1da5de20c7","subject":"Merge \"Encrypted Emulated Virtual TPM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b259a414ae6f12175f20de5318e2df1da5de20c7"}]}],"author":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 03:34:48.000000000","tz":480},"committer":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 03:34:48.000000000","tz":480},"subject":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec","message":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec\n\nChange-Id: I96528ae95a8914431850039c93c4a24b303f73ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6900909a45b880ebec9ae9d72f43268b7a9e369a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6900909a45b880ebec9ae9d72f43268b7a9e369a"}]},"branch":"refs/heads/master"},"890787bd4fbcbb3fbf527d5de7c653550f966cbf":{"kind":"REWORK","_number":2,"created":"2020-06-18 05:07:19.000000000","uploader":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"ref":"refs/changes/22/736422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/22/736422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/2"}}},"commit":{"parents":[{"commit":"b259a414ae6f12175f20de5318e2df1da5de20c7","subject":"Merge \"Encrypted Emulated Virtual TPM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b259a414ae6f12175f20de5318e2df1da5de20c7"}]}],"author":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 03:34:48.000000000","tz":480},"committer":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 05:05:17.000000000","tz":480},"subject":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec","message":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec\n\nChange-Id: I96528ae95a8914431850039c93c4a24b303f73ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/890787bd4fbcbb3fbf527d5de7c653550f966cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/890787bd4fbcbb3fbf527d5de7c653550f966cbf"}]},"branch":"refs/heads/master"},"811e4b42559ec3cfb17b36d8f79b7c7fc16dff99":{"kind":"REWORK","_number":3,"created":"2020-06-18 05:43:45.000000000","uploader":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"ref":"refs/changes/22/736422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/22/736422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/22/736422/3"}}},"commit":{"parents":[{"commit":"b259a414ae6f12175f20de5318e2df1da5de20c7","subject":"Merge \"Encrypted Emulated Virtual TPM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b259a414ae6f12175f20de5318e2df1da5de20c7"}]}],"author":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 03:34:48.000000000","tz":480},"committer":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 05:42:24.000000000","tz":480},"subject":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec fix document error","message":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec\nfix document error\n\nChange-Id: I96528ae95a8914431850039c93c4a24b303f73ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/811e4b42559ec3cfb17b36d8f79b7c7fc16dff99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/811e4b42559ec3cfb17b36d8f79b7c7fc16dff99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
