)]}'
{"id":"openstack%2Fnova-specs~736425","triplet_id":"openstack%2Fnova-specs~master~I9ef62cd9d4cd7d1dd2d6f3c50b14a9ba462aced3","project":"openstack/nova-specs","branch":"master","topic":"nova-support-multiple-boot-volume-with-boot-order-selection","hashtags":[],"change_id":"I9ef62cd9d4cd7d1dd2d6f3c50b14a9ba462aced3","subject":"fix \"Line limited\" error","status":"ABANDONED","created":"2020-06-18 04:05:11.000000000","updated":"2020-06-18 05:08:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c3f063d15b29612e9b2932e9fd341a6057f36127","_number":736425,"virtual_id_number":736425,"owner":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-18 04:10:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 04:10:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f367200f2d8866d4ff99575655a0947346a5a9ec","author":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"date":"2020-06-18 04:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8614d9458a7c4b37008921084948caf95b0ba191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 04:10:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b0a28534ae4cfda84fbb3ba8c68f4f : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8826da8a1bb247aea93d7cb83bd5cd9b : SUCCESS in 4m 40s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/rocky/approved/nova-support-multiple-boot-volume-with-boot-order-selection.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/rocky/approved/nova-support-multiple-boot-volume-with-boot-order-selection.rst","accounts_in_message":[],"_revision_number":1},{"id":"28192d3b3690af1e965cb3bf4a2fece1461e42d0","author":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"date":"2020-06-18 05:08:31.000000000","message":"Abandoned\n\ncommitted by mistake","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f03f5fa8adf9a117e4ba728c33ce648ca4eea08","revisions":{"6f03f5fa8adf9a117e4ba728c33ce648ca4eea08":{"kind":"REWORK","_number":1,"created":"2020-06-18 04:05:11.000000000","uploader":{"_account_id":32039,"name":"Jinsheng Zhang","email":"zhangjl02@inspur.com","username":"zhangjl02"},"ref":"refs/changes/25/736425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/25/736425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/25/736425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/25/736425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/25/736425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/25/736425/1"}}},"commit":{"parents":[{"commit":"6900909a45b880ebec9ae9d72f43268b7a9e369a","subject":"Add nova-support-multiple-boot-volume-with-boot-order-selection spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6900909a45b880ebec9ae9d72f43268b7a9e369a"}]}],"author":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 04:02:11.000000000","tz":480},"committer":{"name":"zhangjl02","email":"zhangjl02@inspur.com","date":"2020-06-18 04:03:31.000000000","tz":480},"subject":" fix \"Line limited\" error","message":" fix \"Line limited\" error\n\nChange-Id: I9ef62cd9d4cd7d1dd2d6f3c50b14a9ba462aced3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6f03f5fa8adf9a117e4ba728c33ce648ca4eea08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6f03f5fa8adf9a117e4ba728c33ce648ca4eea08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
