)]}'
{"id":"openstack%2Fnova-specs~739349","triplet_id":"openstack%2Fnova-specs~master~Ie9e53ea700d933a27837d0a6408af101955ae12a","project":"openstack/nova-specs","branch":"master","topic":"bp/volume-backed-server-rebuild","hashtags":[],"change_id":"Ie9e53ea700d933a27837d0a6408af101955ae12a","subject":"add volume backed server rebuild spec","status":"ABANDONED","created":"2020-07-05 09:33:02.000000000","updated":"2021-12-14 08:47:13.000000000","total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d4ca985dbb3a59d0e675f52f8b7431d28c378252","_number":739349,"virtual_id_number":739349,"owner":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"date":"2020-08-04 18:04:31.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"date":"2020-08-13 05:06:07.000000000","_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-02 08:39:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-1,"date":"2020-08-13 13:33:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"date":"2020-12-01 13:28:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-08-04 18:58:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2020-08-10 03:46:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"value":-1,"date":"2020-10-21 14:23:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2020-11-09 14:30:11.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-05 09:34:24.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2020-07-05 09:35:59.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-07-05 09:37:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-07 06:52:32.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-07-07 11:47:59.000000000","updated_by":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"reviewer":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"state":"REVIEWER"},{"updated":"2020-07-07 11:48:07.000000000","updated_by":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"reviewer":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"state":"REVIEWER"},{"updated":"2020-07-08 00:14:42.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-08-02 08:39:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 18:04:31.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-08-04 18:58:52.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-08-06 17:31:41.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-08-10 03:46:48.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2020-08-13 13:33:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-10-21 14:23:47.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-11-09 14:30:11.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-03-02 07:22:59.000000000","updated_by":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"reviewer":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"state":"CC"},{"updated":"2021-03-02 07:23:08.000000000","updated_by":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"reviewer":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"state":"REMOVED"},{"updated":"2021-09-03 08:39:13.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2021-12-14 08:47:13.000000000","updated_by":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"reviewer":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"state":"CC"}],"messages":[{"id":"ec240914f37b6d54b9f3a8e520735c20b5050fe8","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-05 09:33:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5fdf1608d323182ea9c4f418023678cd7bb8b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 09:41:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb4950fda424d45ab58c392ce0e9032 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c3015179d446e59d13bd193effdde2 : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5ce462dd0a9018d6ebfb87d5b02cd3af60d9c5af","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-06 03:10:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a666c26c3d956ffc95dbfb1db00359e4d5b7d60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 03:20:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20422eb57bf4cb59737368b3ac44284 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd522f63b99c467ab0a12430a23928b4 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ebc6c5b47c52474bc87ce542abbb95683646d2c1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-07 02:09:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7305eec5c3404bf128207c61a96258fb09153df6","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-07 03:57:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"caf2f07e90670acdb7faf0d65a6963f6e0bcc71e","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-07 03:58:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"766dfe99811be83b718b88ab388dcc47b232fb98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 04:08:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1043995e48419d80f17756de4dcc51 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba771a450d0440dacef42cd6a6af73d : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3ea3cf14cd129b1a8ba2aa87e464286772c4dbf2","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-07-07 06:52:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d98c4f0b015362f34c6e5f2aa10936d6d02bf8f0","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-07 12:07:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e14e2075cd755eb5e2b983fd4ddee017e16aa77","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-07 12:08:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"edf8aa06650f30a257d3ba57a6076cd175e16db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 12:17:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2ef4484d134c4d8340f9653b4bd4db : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d0b6f3f57844152953245e9f94de2b6 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"37c16d9b748de272dbc87f5a028f8e2be54e7c9e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-08 00:14:42.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e06470c17d5896dee020794a64e6255a1631c402","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-08 03:22:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"36dc3393499718ec70691acb6d7dc22fb2b7cb82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 03:30:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c76f5efeaf4d2087a75e59a024ac9e : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c1a87c994454ffd82a932ef7a49fd24 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":5},{"id":"96b09f0ac131695732228250f41fb7c78b7b85a4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-08 15:08:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)\n\nLooks good to me. If you need to respin the spec then please fix my inline nits.","accounts_in_message":[],"_revision_number":5},{"id":"c751b77b56c3398ecb9535ccb4db43da4ec68e5a","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-09 07:55:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"527599c9154860b2c9e438fc04646ef2e348b9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:08:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f18724753643908b3f67efb629c2a0 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c94a6194019484d9fae932ae5279ae6 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":6},{"id":"638f17aaf4d68f3281acfec62d574744d61757c7","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-09 15:19:52.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)\n\nA good start, I\u0027d like to see more details about the Cinder API and for us to potentially use traits instead of a RPC bump to indicate support here.","accounts_in_message":[],"_revision_number":6},{"id":"c30ba855e071ecbf6d2dbca88d9440ff35b9f38c","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-15 06:21:42.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9c1bd838ee7a627725f55ae97ff0795c8497edd6","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-15 06:22:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5d9af1619a204f2f535ba9e5e01ccdad8eaa42e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 06:35:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a36961f58446a5abc4cf0291f23c6f : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713cdefe47aa4c0a93211d5e92644d96 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":7},{"id":"d221b02aa72282cfd1a5ec37d3fe52603654b879","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-07-21 12:48:13.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"14bbac76d8a672ff7469f7e27dff493c9d267c84","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-30 08:59:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3359945a202b3fb4116009998bb48df7d5d617e7","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-30 09:02:18.000000000","message":"Patch Set 8:\n\n\u003e (1 comment)\n\nThanks Lee and Sylvain.","accounts_in_message":[],"_revision_number":8},{"id":"5bb08f612ccfd1042bd83df2b7aa27c1ed4771a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 09:09:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39c59fc9a464ae4b6022b3726b40894 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ddb38fb39a4fccb25c7c8870053986 : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":8},{"id":"53d474236e2a7625779074bb4e39e3e0bee59127","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-07-30 15:23:39.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f8c889b5d86928e1f75f6db7efdcc5b6a972db5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 15:38:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b2ec05284e4861911ec5f8cb8e3833 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54021148ed174395b3956efaba977112 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":9},{"id":"00844b07b94959e8f9e5e5429b167208a03cf3b4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-31 11:04:01.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)\n\nthis is a soft -1 the only thing i really would like to be changed is using an external event rather than polling","accounts_in_message":[],"_revision_number":9},{"id":"1f6f542b48259a8285e8e6e98a9d1e904989d594","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-02 08:04:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"56a8d1fb93d0a5562f9906445787dfaddedef35e","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-02 08:09:00.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9476876e528d46c714321d4fe30305af75b050ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 08:16:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a7261076c74dc8bb04b75849fd3473 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2645d1a87e4bc894cb74c5783dbf45 : SUCCESS in 3m 41s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/victoria/approved/volume-backed-server-rebuild.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/victoria/approved/volume-backed-server-rebuild.rst","accounts_in_message":[],"_revision_number":10},{"id":"f225a4e55a96ea2a619b6edce5b86963781a5bc5","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-02 08:31:10.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"092a4e51f9d3c163e9f1735cd502901cdfad236b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 08:39:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc1fd015661487ab65189abdb5e00d4 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd9ee2f11c24ecc9bffd4994ed9de74 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":11},{"id":"d78138509bfc22339a8c01c72280dca6a2d60c39","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-03 11:28:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ca3cd59b960b75f191fede2a45dad114302891f3","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-03 14:10:06.000000000","message":"Patch Set 11:\n\n\u003e (1 comment)\n\nNot yet， I will join the Cinder weekly meeting to communicate with Cinder team in this Wednesday.","accounts_in_message":[],"_revision_number":11},{"id":"f3da1127379b02fcc04c03555b2950405b2e1c4a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-03 22:10:04.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9c3efd09439ff123950488b65a89e33628fdb80d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 18:04:31.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c9db8fb9064dc53fa7ffe6025d0cac31231b0239","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-04 18:58:52.000000000","message":"Patch Set 11: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8b4ff5728feb3284ddd436e71f7fc656e26b6ba5","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-06 14:02:47.000000000","message":"Patch Set 11:\n\nI communicated with the Cinder team, there are some questions to determine, so they will discuss this more at the midcycle next week.\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-meeting-alt/%23openstack-meeting-alt.2020-08-05.log.html#t2020-08-05T14:25:16","accounts_in_message":[],"_revision_number":11},{"id":"0a06cb81b93f137d4f4b3c20e83748aca31e40ad","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-06 17:31:41.000000000","message":"Patch Set 11: Code-Review+2\n\nAs we discussed on the today on the nova meeting if Cinder is OK with the addition of the os-server-external-event sending then I\u0027m OK to grant freeze exception for this spec. However the deadline to get the spec approved is end of this week.","accounts_in_message":[],"_revision_number":11},{"id":"4a4e220621075c52398105e53be4b2c981041406","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-07 01:27:59.000000000","message":"Patch Set 11:\n\nThanks all, the Cinder team can\u0027t decide it this week. I will repropose this in Wallaby.","accounts_in_message":[],"_revision_number":11},{"id":"a00cb3e12b8d5b645e91b4d9ffd9b051e82cb9ab","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2020-08-10 03:46:48.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"454e8bc7bb5b725d4cd373d61b41e7c190c880d2","author":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"date":"2020-08-13 05:06:07.000000000","message":"Patch Set 11:\n\nthe Cinder team is aprroved to use external event:https://etherpad.opendev.org/p/cinder-victoria-mid-cycles","accounts_in_message":[],"_revision_number":11},{"id":"f66340ba936be0228179a9bb0ab337838a91c151","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-13 13:33:59.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nJust want to flag a clarification that will have to be made in the Compute API ref.","accounts_in_message":[],"_revision_number":11},{"id":"7651d629e1fd274df501ee7bd4bcc93cc1b4a592","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-10-21 14:23:47.000000000","message":"Patch Set 11: Code-Review-1\n\n(4 comments)\n\nmost of my concerns go to the upgrade path, and what to do in case of an old Cinder API.","accounts_in_message":[],"_revision_number":11},{"id":"70dc4304cffdce23af96f6cb5f0f68a3f1cf02b8","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-11-09 14:30:11.000000000","message":"Patch Set 11: Workflow-1\n\nNeeds to be moved to specs/wallaby.","accounts_in_message":[],"_revision_number":11},{"id":"ac637673e3215ae2c175e7ce53342afe2f804b82","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-01 13:28:22.000000000","message":"Patch Set 11: -Code-Review\n\nping: are you still pushing this spec for wallaby?","accounts_in_message":[],"_revision_number":11},{"id":"ea9c595f513c9a45fad570be8024b24892044622","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2021-03-02 07:23:08.000000000","message":"Removed reviewer Jorhson Deng.","accounts_in_message":[],"_revision_number":11},{"id":"c75fbb5e0d3ef5a1ab335407726446153e9ce4a7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 12:21:14.000000000","message":"Abandoned\n\nLast response from the author was more than 6 months ago. So I\u0027m abandoning this spec. Feel free to restore it or ping me (gibi) on IRC to do so","accounts_in_message":[],"_revision_number":11},{"id":"911c9e64a3e994464d3205851fa82b93988993a9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-03 08:39:13.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nI\u0027ve added this to the yoga PTG etherpad for cinder[1].\nA cross project session would be good for this, If this feature still makes sense for nova-cinder usecase.\n\n[1] https://etherpad.opendev.org/p/yoga-ptg-cinder-planning","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"5ed946d72d70e5ed9aca74383a6d1bb26304bb2d","revisions":{"fd065f60e80cb75e8efae92d8ef2e54e8e59f495":{"kind":"REWORK","_number":1,"created":"2020-07-05 09:33:02.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/1"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:32:42.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nCo-Authored-By: Kevin_Zheng \u003czhengzhenyu@huawei.com\u003e\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fd065f60e80cb75e8efae92d8ef2e54e8e59f495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fd065f60e80cb75e8efae92d8ef2e54e8e59f495"}]},"branch":"refs/heads/master"},"a9d1b01268fabccb9c38a6344d85883422833687":{"kind":"REWORK","_number":2,"created":"2020-07-06 03:10:04.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/2"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-06 03:09:09.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nCo-Authored-By: Kevin_Zheng \u003czhengzhenyu@huawei.com\u003e\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a9d1b01268fabccb9c38a6344d85883422833687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a9d1b01268fabccb9c38a6344d85883422833687"}]},"branch":"refs/heads/master"},"816f3e2872bbb0781ac7769ef01df5843f6efd74":{"kind":"REWORK","_number":3,"created":"2020-07-07 03:58:18.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/3"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-07 03:57:15.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/816f3e2872bbb0781ac7769ef01df5843f6efd74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/816f3e2872bbb0781ac7769ef01df5843f6efd74"}]},"branch":"refs/heads/master"},"50fff06aca14a893ea54b4194a4cc7931a39d5f4":{"kind":"REWORK","_number":4,"created":"2020-07-07 12:07:17.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/4"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-07 12:06:39.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/50fff06aca14a893ea54b4194a4cc7931a39d5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/50fff06aca14a893ea54b4194a4cc7931a39d5f4"}]},"branch":"refs/heads/master"},"3de97d76a254742a150ca69c1b26f5b431275d5c":{"kind":"REWORK","_number":5,"created":"2020-07-08 03:22:26.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/5"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-08 03:21:55.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3de97d76a254742a150ca69c1b26f5b431275d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3de97d76a254742a150ca69c1b26f5b431275d5c"}]},"branch":"refs/heads/master"},"d240cc9a51e8aeacd5d42bfdf38011d28a38646f":{"kind":"REWORK","_number":6,"created":"2020-07-09 07:55:48.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/6"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-09 07:54:35.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d240cc9a51e8aeacd5d42bfdf38011d28a38646f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d240cc9a51e8aeacd5d42bfdf38011d28a38646f"}]},"branch":"refs/heads/master"},"3eeaff4aae173674fd7f7c73df471b106ca67219":{"kind":"REWORK","_number":7,"created":"2020-07-15 06:22:49.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/7"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-15 06:21:22.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3eeaff4aae173674fd7f7c73df471b106ca67219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3eeaff4aae173674fd7f7c73df471b106ca67219"}]},"branch":"refs/heads/master"},"2b67519b5ecdb787bb182eff0c58cf1745c41538":{"kind":"REWORK","_number":8,"created":"2020-07-30 08:59:41.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/8"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-30 08:58:21.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b67519b5ecdb787bb182eff0c58cf1745c41538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b67519b5ecdb787bb182eff0c58cf1745c41538"}]},"branch":"refs/heads/master"},"c622aaed4815ccefcfe3a558a3d634c7a4a2e599":{"kind":"REWORK","_number":9,"created":"2020-07-30 15:23:39.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/9"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-30 15:21:48.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c622aaed4815ccefcfe3a558a3d634c7a4a2e599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c622aaed4815ccefcfe3a558a3d634c7a4a2e599"}]},"branch":"refs/heads/master"},"64dbbc9efb08a19eea696ee25393a49685737900":{"kind":"REWORK","_number":10,"created":"2020-08-02 08:04:56.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/10"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-08-02 08:04:11.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64dbbc9efb08a19eea696ee25393a49685737900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/64dbbc9efb08a19eea696ee25393a49685737900"}]},"branch":"refs/heads/master"},"5ed946d72d70e5ed9aca74383a6d1bb26304bb2d":{"kind":"REWORK","_number":11,"created":"2020-08-02 08:31:10.000000000","uploader":{"_account_id":27594,"name":"Jie Li","email":"lijie@unitedstack.com","username":"ramboman"},"ref":"refs/changes/49/739349/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/49/739349/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/49/739349/11"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-07-05 09:26:47.000000000","tz":480},"committer":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-08-02 08:29:51.000000000","tz":480},"subject":"add volume backed server rebuild spec","message":"add volume backed server rebuild spec\n\nThe spec for support rebuild volume backed\nserver was accepted in Stein, but some of\nthe work did not finish, so repropose it\nfor Victoria.\n\nblueprint volume-backed-server-rebuild\n\nPreviously-approved: Stein\n\nChange-Id: Ie9e53ea700d933a27837d0a6408af101955ae12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5ed946d72d70e5ed9aca74383a6d1bb26304bb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5ed946d72d70e5ed9aca74383a6d1bb26304bb2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
