)]}'
{"id":"openstack%2Fnova-specs~740863","triplet_id":"openstack%2Fnova-specs~master~I31795585615ee24d029e10538061f313895d76c3","project":"openstack/nova-specs","branch":"master","hashtags":[],"change_id":"I31795585615ee24d029e10538061f313895d76c3","subject":"migrate testing to ubuntu focal","status":"ABANDONED","created":"2020-07-14 03:43:05.000000000","updated":"2020-08-07 00:46:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"89be5e74543147ebb3ab354242f727dfede16770","_number":740863,"virtual_id_number":740863,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"date":"2020-07-16 07:49:40.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-16 08:03:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2020-07-30 23:13:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":-1,"date":"2020-07-20 16:27:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 06:41:24.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-15 11:31:41.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-07-16 07:49:40.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-07-16 08:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-20 16:27:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-07-30 23:13:55.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"b92d85f68c64b1ba3f651494787043039959e961","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-07-14 03:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11567d3d475ff7d5f74810df9ed7110538776f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 03:52:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98de4fa2808457192edf1510403eb0c : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb84b1eb3e14a7c9b4a4546ac397383 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"451f170365687d194ff336b6ee6a2eae472ea419","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-14 06:41:24.000000000","message":"Patch Set 1: Code-Review-1\n\nyour commit message is not describing the change, please update. There\u0027s nothing here that goes to testing focal.\n\nAlso, update directly to hacking 3.2.0.","accounts_in_message":[],"_revision_number":1},{"id":"739f48d83d62dbe0c070af2c44824d011011fe76","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-07-15 11:31:41.000000000","message":"Patch Set 1: Code-Review-1\n\nyour commit message does not match the contents. please update your commit message,and hacking 3.2.0 is not out yet, could you provide more infomation?","accounts_in_message":[],"_revision_number":1},{"id":"9ed1dba659847c8adf1f24eb142ca161fb86d111","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-16 07:45:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"71450f6c5b40886f54f4d46ba454539b90a77508","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-16 07:49:40.000000000","message":"Patch Set 2:\n\n\u003e Also, update directly to hacking 3.2.0.\n\nTo quote gmann: \"i am not saying hacking 3.2.0 hit this issue. I am saying pyflakes can be removed when this repo start using hacking 3.2.0\"","accounts_in_message":[],"_revision_number":2},{"id":"fce143a616bb62251008dc33dc52280bc62395e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:03:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea563affe2aa466088f82f1b83ef00de : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8699cdaaaa4813964e1e866b8d613d : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d6f18f143077e3ce12d7b3f46568e291523a0fec","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-20 16:27:37.000000000","message":"Patch Set 2: Code-Review-1\n\nwe need to test this first if it is failing or not https://review.opendev.org/#/c/741984/","accounts_in_message":[],"_revision_number":2},{"id":"a4ea297c061c89512db5b384848a808ecbef2953","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-20 16:40:20.000000000","message":"Patch Set 2:\n\nno work needed here, working fine on focal -https://review.opendev.org/#/c/741984/","accounts_in_message":[],"_revision_number":2},{"id":"7c55733ecfef7aadc43716ef895514dc2467be2d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-30 23:13:55.000000000","message":"Patch Set 2: Code-Review-1\n\nthis does not need a spec.\nthis will be done anyway as part of \nhttps://opendev.org/openstack/governance/src/branch/master/reference/runtimes/victoria.rst\n\nso this should be abandoned.","accounts_in_message":[],"_revision_number":2},{"id":"aa59427e9b9666e8e58e2014b801153407760f1a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-30 23:16:30.000000000","message":"Patch Set 2:\n\noh missread the commit i taught this was about move nova to use focal for testing.\nim obviously tired :)\n\nbut as gmann has show in https://review.opendev.org/#/c/741984/\nit passes without this.","accounts_in_message":[],"_revision_number":2},{"id":"89be5e74543147ebb3ab354242f727dfede16770","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-08-07 00:46:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59fbb5a9118a3eb1e546159e18a367b8a7d5970c","revisions":{"aec8133a66f31b3f6b444a78f54bb80d036b5259":{"kind":"REWORK","_number":1,"created":"2020-07-14 03:43:05.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/63/740863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/740863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/1"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-07-14 03:42:37.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-07-14 03:42:37.000000000","tz":480},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nChange-Id: I31795585615ee24d029e10538061f313895d76c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aec8133a66f31b3f6b444a78f54bb80d036b5259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aec8133a66f31b3f6b444a78f54bb80d036b5259"}]},"branch":"refs/heads/master"},"59fbb5a9118a3eb1e546159e18a367b8a7d5970c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-16 07:45:30.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/63/740863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/740863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/740863/2"}}},"commit":{"parents":[{"commit":"619854e6dedeca362b4a15bb8d9efad866f70da6","subject":"Merge \"Update the testing section of the rbd-glance-multistore spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/619854e6dedeca362b4a15bb8d9efad866f70da6"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-07-14 03:42:37.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-07-16 07:44:46.000000000","tz":0},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nStory: #2007865\nTask: #40178\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I31795585615ee24d029e10538061f313895d76c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/59fbb5a9118a3eb1e546159e18a367b8a7d5970c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/59fbb5a9118a3eb1e546159e18a367b8a7d5970c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
