)]}'
{"id":"openstack%2Fnova-specs~742785","triplet_id":"openstack%2Fnova-specs~master~Ia425629ae5222d6fa7485845533231d0c04a989b","project":"openstack/nova-specs","branch":"master","topic":"bp/sriov-smartnic-support","hashtags":[],"change_id":"Ia425629ae5222d6fa7485845533231d0c04a989b","subject":"Smart-nic Support Specification","status":"MERGED","created":"2020-07-24 03:15:58.000000000","updated":"2020-12-14 02:53:40.000000000","submitted":"2020-12-14 02:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":188,"unresolved_comment_count":32,"has_review_started":true,"submission_id":"742785-bp/sriov-smartnic-support","meta_rev_id":"e2315801fc211a08e70900175770918f9a86eca9","_number":742785,"virtual_id_number":742785,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-14 02:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-11 16:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":1,"date":"2020-12-11 16:30:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-12-10 08:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-12-14 02:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-12-14 02:37:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 04:16:55.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-07-24 04:17:35.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2020-07-28 04:21:56.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-07-30 23:00:56.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-07-31 11:03:53.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-08-06 03:10:55.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-10-05 12:52:09.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-10-21 02:54:49.000000000","updated_by":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"reviewer":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"state":"REVIEWER"},{"updated":"2020-10-28 11:55:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-17 11:59:54.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-11-17 14:15:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-18 07:21:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-18 07:22:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-11-18 07:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-09 12:21:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"}],"messages":[{"id":"efcb6fb34c3b3a5703d14d0598bb679415c2002c","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-24 03:15:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56b9e816a7ffb66d6b75224f2ea3cbb547619941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:24:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1388576255b149f29a60abe748e95777 : FAILURE in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6f3a4b1c614d2a88485126a6c02a2f : SUCCESS in 4m 25s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/victoria/approved/support-sriov-smartnic.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/victoria/approved/support-sriov-smartnic.rst","accounts_in_message":[],"_revision_number":1},{"id":"1553dfbbdba7f08bc1a0866ed519923f168f0f0c","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-24 04:16:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c7803c220c66366858beda58149d189f4499300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 04:30:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca8f1a923a5496e86a0dda7e1f0f3e5 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c7fa7996ac4b59af9775ecf62c2e25 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"721f61f49f18dfdd57ad64f4481202d3402190aa","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-07-25 03:06:48.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4630485f430190aedfbc5395bb110d6f65d960d1","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-07-27 07:17:34.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"607e7b328340ccb16ddb44e4a1c7624142a83aa8","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-07-28 04:21:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1cf3582df0a42281db3d26e9dbf292ce7b39edb6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-30 23:00:56.000000000","message":"Patch Set 2: Code-Review-1\n\ni have not had time to review this tonight\nhowever i think given were we are in the cycle we likely shoudl defer this to W. we and a number of openquestions after the ptg and unless they have all been adress in this spec which is suspect is not the case i dont think we can realisticly resolve this in the time we ahve left.\n\nthat does not mean the cyborg work cant continute but i think it would be better to focus on completeding the lifecycle operation for cyborg instances.\nhttps://review.opendev.org/#/c/715326/\nhttps://review.opendev.org/#/c/729945/\nhttps://review.opendev.org/#/c/729563\n\nbefore starting smart nic integration and also focus on testing as the current cyborg tempest job does not properly test cyborg integration.","accounts_in_message":[],"_revision_number":2},{"id":"69e6f0918dac30fa5b641088a3b52cd8782c8062","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-07-31 11:03:53.000000000","message":"Patch Set 2:\n\n\u003e i have not had time to review this tonight\n \u003e however i think given were we are in the cycle we likely shoudl\n \u003e defer this to W. we and a number of openquestions after the ptg and\n \u003e unless they have all been adress in this spec which is suspect is\n \u003e not the case i dont think we can realisticly resolve this in the\n \u003e time we ahve left.\n \u003e \n \u003e that does not mean the cyborg work cant continute but i think it\n \u003e would be better to focus on completeding the lifecycle operation\n \u003e for cyborg instances.\n \u003e https://review.opendev.org/#/c/715326/\n \u003e https://review.opendev.org/#/c/729945/\n \u003e https://review.opendev.org/#/c/729563\n \u003e \n \u003e before starting smart nic integration and also focus on testing as\n \u003e the current cyborg tempest job does not properly test cyborg\n \u003e integration.\n\n \u003e i have not had time to review this tonight\n \u003e however i think given were we are in the cycle we likely shoudl\n \u003e defer this to W. we and a number of openquestions after the ptg and\n \u003e unless they have all been adress in this spec which is suspect is\n \u003e not the case i dont think we can realisticly resolve this in the\n \u003e time we ahve left.\n \u003e \n \u003e that does not mean the cyborg work cant continute but i think it\n \u003e would be better to focus on completeding the lifecycle operation\n \u003e for cyborg instances.\n \u003e https://review.opendev.org/#/c/715326/\n \u003e https://review.opendev.org/#/c/729945/\n \u003e https://review.opendev.org/#/c/729563\n \u003e \n \u003e before starting smart nic integration and also focus on testing as\n \u003e the current cyborg tempest job does not properly test cyborg\n \u003e integration.\n\nHi Sean, Thanks for sharing your ideas!\n\n\"lifecycle operations for cyborg instances\" are one of cyborg\u0027s high priority goals in V release, we should focus on them.\nBrin, wenping and you are contributing on them, thanks for the efforts! I will make time to review them all and keep constant attention.\nand if there\u0027s any other related patches,pls add me to the reviewer, or if any other efforts are needed pls also feel free to ping me.\n\nfor the smartnic spec, you know it takes long time to get ideas from different aspects, hope we can have progress in this release, reaching closer to an possible agreement.","accounts_in_message":[],"_revision_number":2},{"id":"8dc541a1f3a42a6687097f85a657d8e06620bd36","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-31 11:10:54.000000000","message":"Patch Set 2:\n\nwe can continue to work on the spec i would just not expect us to reach an agreement by next week. technically we are past the spec approval deadline now we may appove some addtional spec that already have overall agreement but i dont think this would qualify as an excpetion since there are still open questions to be resolved. that does not perclude us still working on that just the code and spec will likely land in W instead of victoria.","accounts_in_message":[],"_revision_number":2},{"id":"ce7c95c99bc05126988a71795678b710316d6cf2","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-08-03 02:19:21.000000000","message":"Patch Set 2:\n\n\u003e we can continue to work on the spec i would just not expect us to\n \u003e reach an agreement by next week. technically we are past the spec\n \u003e approval deadline now we may appove some addtional spec that\n \u003e already have overall agreement but i dont think this would qualify\n \u003e as an excpetion since there are still open questions to be\n \u003e resolved. that does not perclude us still working on that just the\n \u003e code and spec will likely land in W instead of victoria.\n\nYes, sure. There are still open questions need to be nailed down, this spec is a start, we can continue working on this, discuss here in order to reach an agreement.","accounts_in_message":[],"_revision_number":2},{"id":"7e186db22e91c14604625aaba190f7f48c01a793","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-08-10 06:52:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98c4c76816e5e91a2d4f0335165e10507022c958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 07:06:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edab97a053ce4592b4a1a239d8438bb3 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868d8e6848054bc793110845c31da0fc : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"fd4e8ef2bd5f6cf3f6931799fb903bab07ab6407","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-08-28 06:43:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba9bc65046ecab5ba33bf31862ef6f2632e6bf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:02:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58117c2779994202999d4972f7976057 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60484f88db7c4afda5ad8e0a254cbc02 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":4},{"id":"d7ae73c618a84bed09563daf48e841bb5f508882","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-10-01 15:59:57.000000000","message":"Patch Set 4: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"20e2b8cb3675837f466d38d9ff22e02e975d4777","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-10-21 02:54:49.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ad7c95927bb65f70c7f479aecba41c9be9d8c49a","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-24 06:12:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"7891ebe6a50b3cac00aa25fafb329f4db44b8fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 06:27:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84704124c2e496abb8009b3151451fa : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64fcf3702b9c4af492b7bf137e1921c8 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":5},{"id":"2280e275c112cb97f51c480a52798aa5c05c4a26","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-10-30 07:30:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"949f8d627283ed6447634cdde0ce7c283bb61f0f","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-10-30 07:31:42.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cfa96a23fdf2d1079e1a2159fb536b7f0c7ed81e","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-10-30 07:33:34.000000000","message":"Topic changed from sriov-nic-support to bp/sriov-smartnic-support","accounts_in_message":[],"_revision_number":6},{"id":"3bff4cb75f044622c5641c14aab672635ad802b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:56:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a87453255348de882a40409fe9754a : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d1a7e3363a4930944e2f806da5cca0 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":6},{"id":"b141a4d35e383a48e2cff34ea15994e62a931c82","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-10-30 10:22:42.000000000","message":"Patch Set 6: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f3ab77e540ac2f0af544ff6fa4319314418f7bf1","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-03 02:37:58.000000000","message":"Patch Set 6:\n\n(7 comments)\n\nThanks for your comments. Please see my reply inline.","accounts_in_message":[],"_revision_number":6},{"id":"86b87563f1982623c15b9f2149671a530d1a324d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-05 13:22:12.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8944be4f1c34651b245ec0132ac08a612fe71470","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-10 06:28:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b6aba35b8ed8a3eadb92134c11146cab831b2898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 06:46:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fa522786be4e098f607d8b15ab77fe : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809eab1bd99947a9a6a4b14dff3ebfce : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":7},{"id":"ffb8591cac61c63657bd1207111a7840b103513b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-10 07:03:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"abbe6e0d2c1cc885f119cce3fda1b82e8702658b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 07:15:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec903a5c1394a249b9e539308ff97db : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c2d1ca659454aae99ed7c9592e10125 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":8},{"id":"e24f8c0ea34a5864f7d75030aad4d20b1f66bbed","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-10 09:50:14.000000000","message":"Topic changed from sriov-nic-support to bp/sriov-smartnic-support","accounts_in_message":[],"_revision_number":8},{"id":"dfa3f20bab8a70acaf31234913cdbeb20f2155fc","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-12 05:57:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5788c140a52572d86d9da521058c95788c888899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 06:27:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3b570494944afab10dc450423b6a33 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbcfd576ab54551a7d7d4a8e415404b : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":9},{"id":"6e250fd24585cb23f338e705ffcf4469968a45cc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-17 11:59:54.000000000","message":"Patch Set 9: Code-Review-1\n\n(11 comments)\n\nFrom nova perspective I\u0027m OK with the described direction.\n\nFrom process perspective we need acknowledgement from Neutron about the Neutron side changes as well as the close to be approved Cyborg spec. Until that I will keep a procedural -1. (Also I have some formatting and typo nits inline)\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":9},{"id":"cc092a72ad4f5e0b5cb9bc93462376310a9bf3d6","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-18 07:18:20.000000000","message":"Patch Set 9:\n\n(11 comments)\n\nThanks for your review. I have pinged Neutron folks to review this. And some typos are fixed in next commit.","accounts_in_message":[],"_revision_number":9},{"id":"0a39bedfb8441eb457aeba573f7c3a832069dd27","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-18 07:20:41.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4e8fa033045e2f965f43efab41df1ad5955cdc0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 07:38:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e226541a2e04be69080c5a09f8c4942 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d596b0c67d14abc8dbb9907195216b8 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":10},{"id":"0d804e8c47b9f2aa0265878d7547ee8eb003d2b8","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-11-24 06:53:08.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ed2e6c4deb9406844d08f3364509ffb70bdb06f3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-11-24 07:55:45.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fc014c10b1eeeaef5a6da9a1cc6203980d65cdab","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-24 10:45:07.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nThanks for your comments, please see my latest update.","accounts_in_message":[],"_revision_number":10},{"id":"ca9eec1edfdb4e1af390ae8997713404cf63ebe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-24 10:46:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"15f35b91435776c585e0247e96af642aa8cdaf57","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-24 11:01:44.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7fb8636e3e0d755124101cf17e03fb5feb6a17bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 11:02:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d98f3c26b72490288b92f4d30acc866 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c350ed3d78640f6bf6e9be8e7deb8ce : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":11},{"id":"1bdb4fc3474c2bf84fb1095c4599acc45e6abd28","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-24 11:22:46.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1ff19c1af22064b6108be5b05bd55a378526290d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-24 12:37:45.000000000","message":"Patch Set 11: Code-Review-1\n\n(25 comments)","accounts_in_message":[],"_revision_number":11},{"id":"eda4c4eabc1861c6f9db9ccab62f8e334e248ff0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-24 17:37:55.000000000","message":"Patch Set 11: Code-Review-1\n\n(11 comments)\n\nMy main concern is ow do you plan to use port_resource_request to not disturb current QoS minimum_bandwidth use cases.\nPerhaps that was discussed earlier, but can\u0027t figure it out, so forgive me.","accounts_in_message":[],"_revision_number":11},{"id":"b6e6b6b527f395c520cef3b4a60a91a55a5610f3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-25 14:36:31.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nmaintaining my -1 since my comments from","accounts_in_message":[],"_revision_number":11},{"id":"a45f56a1d4558fd80235adc43c52699cdcff73a3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-11-26 05:03:28.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"53914df7c111fcf8b8f0a0668c3639fadfe07493","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-26 09:24:47.000000000","message":"Patch Set 11:\n\n(29 comments)\n\nHi all, thanks for your comments. I have replied some of them. Please check it when you got time. I will continue look into your comments and address them.","accounts_in_message":[],"_revision_number":11},{"id":"312ac2dbb4e7109fbbde0d31e0dec7fbfd4adda0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-26 14:17:44.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nill try and review the latest draft later today\nbut for now just replied to a few things on the previous one.","accounts_in_message":[],"_revision_number":11},{"id":"2db963575f34c883df1084952ca02894848c91e0","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-27 09:37:34.000000000","message":"Patch Set 11:\n\n(5 comments)\n\nPlease see some additional reply here.","accounts_in_message":[],"_revision_number":11},{"id":"40d4c6bee0be7ed7e2e268de19d6a1f7ff121003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-30 03:35:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"09c96b6903d03b78724c91cbb347f6c31e9b08fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 03:42:49.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e23a101b4174454b2e821665397549e : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73869e951d1646de95fa3827544a4d55 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":12},{"id":"f60a4b3ba9e6eb981d8ce812f5586b478da7dfac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-01 12:51:53.000000000","message":"Patch Set 12:\n\nI\u0027m OK with this but other had -1s previously (Alex, Lajos, Sean) so I\u0027m waiting for others to re-review.","accounts_in_message":[],"_revision_number":12},{"id":"dafe18a67c85e8cd614d4eb3be7290baf365b066","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-12-01 13:54:02.000000000","message":"Patch Set 12: Code-Review+1\n\n(7 comments)\n\nsome nits in line but over all im ok with this this as presented.","accounts_in_message":[],"_revision_number":12},{"id":"f2ee09a80649997de688d892903d019cf28369aa","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-12-02 06:22:08.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)\n\nI\u0027m ok with overall flow also. Just some details.","accounts_in_message":[],"_revision_number":12},{"id":"ee650ad61e9d7ad53b1d7d831c24bfe21dda3f88","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-02 07:30:31.000000000","message":"Patch Set 12:\n\nI tend to agree with others that the Neutron part should be discussed on neutron-drivers meeting, and track in an RFE bug or similar","accounts_in_message":[],"_revision_number":12},{"id":"8f2f84f9bc33e5f7ccb2a8378949f76924493de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-12-02 08:15:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a2b114d3b80802c5589c2dd7469210d7314558cb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-12-02 08:18:00.000000000","message":"Patch Set 13: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":13},{"id":"8aa8d677ad597836d9e91dac3a082ff5fb6e7b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 09:00:31.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cec85366d8f4ef1aeb6717314b9ba15 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ed0ff6659848ac9929dc583c35f4d1 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":13},{"id":"edeee8c94885d33ac702d3e1e05b27ee09106b60","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-02 09:27:41.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 12:\n\u003e \n\u003e I tend to agree with others that the Neutron part should be discussed on neutron-drivers meeting, and track in an RFE bug or similar\n\n@Xinran please open an RFE bug on neutron side and refer to it in this spec.","accounts_in_message":[],"_revision_number":13},{"id":"ba80fa4815a3c34dc472374bf23ca85c893a6df1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-12-02 13:26:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"fcbe32fe1e1e5c7b6572e4a0d23fbc217eb6d7c6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-12-02 13:33:35.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b4da5d2ea275a9291c9f4e7bf5340cf01ea52d50","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-12-02 13:36:55.000000000","message":"Patch Set 13:\n\n(7 comments)","accounts_in_message":[],"_revision_number":13},{"id":"1026b550731d2b1cd0681e310bd6c23b0c6ea490","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-12-03 00:50:17.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"29b04175cb0cba5b0a40f462f7a5cc833dbd349b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-12-03 06:56:47.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f7ce2741754a2d588f9d10ae8c7083689a124982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-12-03 07:05:09.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"de983a3b8715d564e518e2170f502d58b8142081","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-03 07:45:46.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)\n\nLGTM.","accounts_in_message":[],"_revision_number":14},{"id":"93cf598dc8f84e722686aea6d9ee7b24c40001e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 08:14:06.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68018b94857e4fe18bdcbadfdae9d83a : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7ef1f1cf194298b522e5e8177b0fd7 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":14},{"id":"8b0ebafa23092fc0c888ad493761e4eb10e76cfa","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-12-04 02:42:26.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2337aa16cdc253af6f2d732a5981ce0a36ad69a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-12-04 02:43:00.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"11fbce1232d6603e64e508dacf2d9a5ce56fc8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 02:57:49.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9cc1953805407d9337ddd31ad1ca46 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352da4a36236486aa0d4d07eb5f185f1 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":15},{"id":"d59dda270700d59a9bd2e5fb47694a9a9605161d","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-12-09 13:39:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d05f84ca00c23592f57e57374fc7bad30236d15e","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-12-10 08:00:51.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"19c5a2f924081ed6d13caf6029991ade54a166c2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-11 16:18:54.000000000","message":"Patch Set 15: Code-Review+2\n\nThe Neutron RFE[1] has been approved. That was my last outstanding thing on this. So I\u0027m +2 now.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1906602","accounts_in_message":[],"_revision_number":15},{"id":"77b9cfd5dc8628751addcfad966b7c6c9096ff0d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-12-11 16:30:32.000000000","message":"Patch Set 15: Code-Review+1\n\ni think this is workable in it current form and we can proceed to the implementation.\nthere may be some detail that come up as this progresses but there is enough detail of the design\ncaptured in this spec to move forward .","accounts_in_message":[],"_revision_number":15},{"id":"51008e0724e928ca767d39af06b44ca9d1958f40","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-12-14 02:37:36.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"4abd880a868fd9cafdcca8ee6654643873f54de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 02:39:27.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"73b5592fafbad5d4a5a7e1f5a9688c73ce1eda11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 02:52:25.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210a1f0d11734786b66bc2b9fe767c97 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3716885371404c10a24e1a9a6f379939 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":15},{"id":"28137a516a6fcaa84a855ac39b96496997777da5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 02:52:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"e2315801fc211a08e70900175770918f9a86eca9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 02:53:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ca54f94f151e418ca0f451e9c349e9f7 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"0eab8bab2dd29bfaa0b7d5e1906752b3d33e8169","revisions":{"6aa5cc070b289a605112023529d46d327a5e9556":{"kind":"REWORK","_number":1,"created":"2020-07-24 03:15:58.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/1"}}},"commit":{"parents":[{"commit":"7ddca43229f1e5bca01389e2bca1e4a595242ec3","subject":"Merge \"Support SRIOV interface attach and detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ddca43229f1e5bca01389e2bca1e4a595242ec3"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:13:02.000000000","tz":0},"subject":"SRIOV SmartNic Support Specification","message":"SRIOV SmartNic Support Specification\n\nThis spec proposes an overall design for smart-nic management which will\ninvolve Nova, Neutron, Cyborg changes. In this spec, we will introduce\nbasically how to manage a smart-nic and how to use a smart-nic for a VM.\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6aa5cc070b289a605112023529d46d327a5e9556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6aa5cc070b289a605112023529d46d327a5e9556"}]},"branch":"refs/heads/master"},"d546540ceea7f8cc175ac6bde01a3917112ee5bf":{"kind":"REWORK","_number":2,"created":"2020-07-24 04:16:01.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/2"}}},"commit":{"parents":[{"commit":"7ddca43229f1e5bca01389e2bca1e4a595242ec3","subject":"Merge \"Support SRIOV interface attach and detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ddca43229f1e5bca01389e2bca1e4a595242ec3"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 04:12:02.000000000","tz":0},"subject":"SRIOV SmartNic Support Specification","message":"SRIOV SmartNic Support Specification\n\nThis spec proposes an overall design for smart-nic management which will\ninvolve Nova, Neutron, Cyborg changes. In this spec, we will introduce\nbasically how to manage a smart-nic and how to use a smart-nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d546540ceea7f8cc175ac6bde01a3917112ee5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d546540ceea7f8cc175ac6bde01a3917112ee5bf"}]},"branch":"refs/heads/master"},"849df9ece79fd9d8c8100e236529326212186127":{"kind":"REWORK","_number":3,"created":"2020-08-10 06:52:49.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/3"}}},"commit":{"parents":[{"commit":"7ddca43229f1e5bca01389e2bca1e4a595242ec3","subject":"Merge \"Support SRIOV interface attach and detach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ddca43229f1e5bca01389e2bca1e4a595242ec3"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-08-10 06:47:25.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/849df9ece79fd9d8c8100e236529326212186127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/849df9ece79fd9d8c8100e236529326212186127"}]},"branch":"refs/heads/master"},"749b12ab19027442f89b18f290e845d93ac36154":{"kind":"REWORK","_number":4,"created":"2020-08-28 06:43:34.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/4"}}},"commit":{"parents":[{"commit":"afab5cf6233efb0fb517f89da03f30b917d54f4d","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/afab5cf6233efb0fb517f89da03f30b917d54f4d"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-08-28 06:27:22.000000000","tz":-240},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/749b12ab19027442f89b18f290e845d93ac36154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/749b12ab19027442f89b18f290e845d93ac36154"}]},"branch":"refs/heads/master"},"35da043ef00f52653635a38f2e0d9ae96e395adf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-24 06:12:15.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/85/742785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/5"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"YumengBao","email":"yumeng_bao@yahoo.com","date":"2020-10-24 06:12:15.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35da043ef00f52653635a38f2e0d9ae96e395adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35da043ef00f52653635a38f2e0d9ae96e395adf"}]},"branch":"refs/heads/master"},"e7d6aa11c190dc6fa2bba29634a7ae17aec13b6e":{"kind":"REWORK","_number":6,"created":"2020-10-30 07:30:47.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/6"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-10-30 07:27:46.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7d6aa11c190dc6fa2bba29634a7ae17aec13b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7d6aa11c190dc6fa2bba29634a7ae17aec13b6e"}]},"branch":"refs/heads/master"},"69215abf825305fc285a24fe12940b3b61970623":{"kind":"REWORK","_number":7,"created":"2020-11-10 06:28:07.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/7"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-11-10 06:23:11.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/69215abf825305fc285a24fe12940b3b61970623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/69215abf825305fc285a24fe12940b3b61970623"}]},"branch":"refs/heads/master"},"ceb702ed94b896530e7916586c07b6c11039368b":{"kind":"REWORK","_number":8,"created":"2020-11-10 07:03:47.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/8"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-11-10 06:59:05.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ceb702ed94b896530e7916586c07b6c11039368b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ceb702ed94b896530e7916586c07b6c11039368b"}]},"branch":"refs/heads/master"},"d69a3bc88a8f50ad929a1fc0dec0e7eb104cef1d":{"kind":"REWORK","_number":9,"created":"2020-11-12 05:57:56.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/9"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-11-12 05:52:23.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d69a3bc88a8f50ad929a1fc0dec0e7eb104cef1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d69a3bc88a8f50ad929a1fc0dec0e7eb104cef1d"}]},"branch":"refs/heads/master"},"1d18eb907bef22f099a03c6352e63ed784b6cb6c":{"kind":"REWORK","_number":10,"created":"2020-11-18 07:20:41.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/10"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-11-18 07:14:51.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d18eb907bef22f099a03c6352e63ed784b6cb6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d18eb907bef22f099a03c6352e63ed784b6cb6c"}]},"branch":"refs/heads/master"},"28b1c5c4a79026fd9be6a7573f716c2876bddadc":{"kind":"REWORK","_number":11,"created":"2020-11-24 10:46:26.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/11"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-11-24 10:40:02.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/28b1c5c4a79026fd9be6a7573f716c2876bddadc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/28b1c5c4a79026fd9be6a7573f716c2876bddadc"}]},"branch":"refs/heads/master"},"744e0b939d34ff8d46271bc7cacf943f62a847b2":{"kind":"REWORK","_number":12,"created":"2020-11-30 03:35:19.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/12"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-11-30 03:27:59.000000000","tz":0},"subject":"SRIOV Enabled Nic Support Specification","message":"SRIOV Enabled Nic Support Specification\n\nThis spec proposes an overall design for sriov enabled nic(including\nsmart-nic) management which will involve Nova, Neutron, Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a sriov enabled\nnic and how to use a sriov enabled nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/744e0b939d34ff8d46271bc7cacf943f62a847b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/744e0b939d34ff8d46271bc7cacf943f62a847b2"}]},"branch":"refs/heads/master"},"e90790d1e327002c1ae6820aaa72467e1b9817ae":{"kind":"REWORK","_number":13,"created":"2020-12-02 08:15:17.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/13"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-12-02 08:07:52.000000000","tz":0},"subject":"Smart-nic Support Specification","message":"Smart-nic Support Specification\n\nThis spec proposes an overall design for smartnic management which\nwill involve Nova, Neutron and Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a smart-nic\nand how to use a smart-nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e90790d1e327002c1ae6820aaa72467e1b9817ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e90790d1e327002c1ae6820aaa72467e1b9817ae"}]},"branch":"refs/heads/master"},"f5aaa6ef56b994811ab02c1308e64f6d98bce604":{"kind":"REWORK","_number":14,"created":"2020-12-03 07:05:09.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/14"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-12-03 06:52:54.000000000","tz":0},"subject":"Smart-nic Support Specification","message":"Smart-nic Support Specification\n\nThis spec proposes an overall design for smartnic management which\nwill involve Nova, Neutron and Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a smart-nic\nand how to use a smart-nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f5aaa6ef56b994811ab02c1308e64f6d98bce604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f5aaa6ef56b994811ab02c1308e64f6d98bce604"}]},"branch":"refs/heads/master"},"0eab8bab2dd29bfaa0b7d5e1906752b3d33e8169":{"kind":"REWORK","_number":15,"created":"2020-12-04 02:43:00.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/85/742785/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/85/742785/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/85/742785/15"}}},"commit":{"parents":[{"commit":"ee8aa085b24384c13987c9ff04deddb09bc29998","subject":"Merge \"Move implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee8aa085b24384c13987c9ff04deddb09bc29998"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-07-24 03:03:32.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-12-04 02:33:37.000000000","tz":0},"subject":"Smart-nic Support Specification","message":"Smart-nic Support Specification\n\nThis spec proposes an overall design for smartnic management which\nwill involve Nova, Neutron and Cyborg changes.\n\nAfter the discusstion during Wallaby PTG, we have a general agreement\nabout the overall design. Please refer to:\nhttps://etherpad.opendev.org/p/nova-wallaby-ptg\n\nIn this spec, we will introduce basically how to manage a smart-nic\nand how to use a smart-nic for a VM.\n\nCo-authored-by: Yongli He\u003cyongli.he@intel.com\u003e\n\nChange-Id: Ia425629ae5222d6fa7485845533231d0c04a989b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0eab8bab2dd29bfaa0b7d5e1906752b3d33e8169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0eab8bab2dd29bfaa0b7d5e1906752b3d33e8169"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
