)]}'
{"id":"openstack%2Fnova-specs~765551","triplet_id":"openstack%2Fnova-specs~master~Ib701e3f79185eabad389136e4411b73d04d8fa04","project":"openstack/nova-specs","branch":"master","topic":"bp/pci-socket-affinity","hashtags":[],"change_id":"Ib701e3f79185eabad389136e4411b73d04d8fa04","subject":"`socket` PCI NUMA-affinity Policy","status":"MERGED","created":"2020-12-04 15:30:55.000000000","updated":"2022-02-03 03:12:48.000000000","submitted":"2021-01-20 11:40:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":78,"unresolved_comment_count":25,"has_review_started":true,"submission_id":"765551-bp/pci-socket-affinity","meta_rev_id":"b41eb505f7067190d84060521e19f52bee9b85c2","_number":765551,"virtual_id_number":765551,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 11:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-01-19 16:44:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-01-20 11:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-02-03 03:12:48.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-20 10:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-01-20 11:29:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 16:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-07 15:31:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-12-07 17:36:11.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2021-01-15 11:16:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-01-19 14:39:40.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-02-03 03:12:48.000000000","updated_by":{"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},"reviewer":{"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},"state":"REVIEWER"}],"messages":[{"id":"139f496efcc4e5cdd671f43aa98ef273979c8c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-12-04 15:30:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b4954fccc21135beb296554f07ab5be6d15c6d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 16:20:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28eb025c7ae24c6087f0f53a4635dd67 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373fe7582746465f904937c5ec7c415d : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"547a39427c22e77299b7321de81fa8e4af534a50","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-07 15:59:34.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nObviously not complete, but some comments inline nonetheless","accounts_in_message":[],"_revision_number":1},{"id":"9a7f5935c8e6dcac75537ebaff22d2c9a60fd664","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-12-07 17:36:11.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nthis is not a full review but just responding to stephens comments.\nill review the full spec later","accounts_in_message":[],"_revision_number":1},{"id":"8a179bb01a120ff1b002f8a7f2f4e8d4973784d1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-12-12 15:49:51.000000000","message":"Topic set to bp/pci-socket-affinity","accounts_in_message":[],"_revision_number":1},{"id":"a4946921e141b925e85a61d41324076dade0034f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-12 21:21:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30183db4e33e692ab07ba08f8b1a727a22030de3","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-12 21:25:29.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"540a34671fb12c75482e2237437269f18f4bb163","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-12 21:26:56.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nBesides my own 2 comments inline, I *think* the other sections are ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"557c174b7b34dd48851bc36b5d7ae28d2115741f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 21:38:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942186d95eed49b899e17537f249becc : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c10cbcc2574d24bfd081b1614a9d85 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6cb1907791e7526640da6670066c035d51ccd2ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-13 19:33:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9b172066ab79a3e420b683bca936fbf3970f8c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:49:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698eb40d70bc4f1fa513f3e25494b251 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940a86da54db43b0aaebaf44851ae4cc : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"f0a8e719f21e4b5bee19dd20052b59187247a3c7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-14 17:56:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)\n\nThis is good. Couple of style nits and one comment on the data model changes. If those can be addressed then I\u0027m +2","accounts_in_message":[],"_revision_number":3},{"id":"06e91e2390522ca0435908ad59ab28f29b08aedc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-14 18:22:20.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2d7d597a9a6f5ce039515ccf1b69573a07cba64d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-01-15 11:16:16.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)\n\njust questions and small comments inline, nothing serious.","accounts_in_message":[],"_revision_number":3},{"id":"c681342667cc7a584f59a6140b3f5330321a08c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-18 19:12:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"40bf14efc581d4037b7bfb70aaf39290ed44fc23","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-18 19:13:19.000000000","message":"Patch Set 4:\n\n(14 comments)\n\nJust Sean\u0027s upgrade question that might need further discussion/adjustements...","accounts_in_message":[],"_revision_number":4},{"id":"c81acd6c1352ccc3282e5a832d31fc13622db843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:22:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9f29e5ce374f94a86b4b682148e82a : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff5baef69aa454ab3485151fe7ae149 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":4},{"id":"e3dbfabc69ad5d75c088f757f568d7e53a5cab71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-19 13:11:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a32c0cefcbdad45aa0d7198d6ea524f02b7f373c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-19 13:11:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"01bf32a0db9cf285f36f4edea62c07e55f0e45af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 13:30:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6712304c9e10448f8c9708722d6f4e05 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98df119d2ec24d3bbd42b4fb87bc8060 : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":5},{"id":"852eebf41f5d7c79e2a6d5bb904770a47868e863","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-01-19 14:02:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9d156515603d5b043e6d040e37c94d724a591af2","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-19 14:28:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30f8cd16e81bc620ea2a7047dc988adfeee3e942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-19 14:33:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3e65095c1ab0f349360ad0c8928e8020a4bdfe29","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-19 14:39:40.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nstill reviwing but just pushing my current comments for you to review","accounts_in_message":[],"_revision_number":6},{"id":"87e8b92f47f51aa4dfbea25bdf7797e42b750558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 14:57:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c81ed6a2f841ab9671efd1031fdfdf : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d261d6d8cd1141ba8a28a6bbe111674a : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":6},{"id":"79ee6d63a6a3fdc4db22168adf5bf9a943c26add","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-19 15:09:40.000000000","message":"Patch Set 6:\n\n(9 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8fd35e40c70fc6bcf20c91467d9e3457231974cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-19 16:05:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a1c183263efcad05f47757f8f82f0fc408a0d088","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-01-19 16:05:23.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b45ff163818168b991bf298b439b0b13aed5648c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 16:24:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5663aaea6e74cb3a15332f4a15597d7 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ffc3ceeaf14dc6a378e1d5a921614e : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":7},{"id":"bc8b6ca54448fee155c9342743a46aff5bc452c8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-19 16:44:17.000000000","message":"Patch Set 7: Code-Review+1\n\ni think im ok with this.\nyou listed my alternative for the data model and added a trait\nas well as updated the example to be  consistent with the existing behaviour for multiple numa nodes.\nso i think my main concerns have been addressed.","accounts_in_message":[],"_revision_number":7},{"id":"16a14ecbdab9474901b29f592132c513111af42d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-01-20 10:29:54.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)\n\nLooks good to me","accounts_in_message":[],"_revision_number":7},{"id":"c59dc0fdaea202277119db9931785e5e275a0665","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-20 11:29:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6b05012bf7cfa2e432846ee174fd7e737be506bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:29:33.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"958c1a2d43792825f9ec22cbfe81711123da0e57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:40:44.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a27dc82db504c1fb4e8275d4cf0f029 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/546c5b8e53294df8ba8c4cff91f6b986 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":7},{"id":"3e5b677a60274377a57557b5c9833d975ae134f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:40:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"986be91d14dfcb502a9d608a66cc1bf361745dbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:43:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/cbfbdd54151a4682aea4d728cef88488 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":7},{"id":"b41eb505f7067190d84060521e19f52bee9b85c2","author":{"_account_id":34537,"name":"Sonemaly Phrasavath","email":"psonemal@amd.com","username":"sphrasavath"},"date":"2022-02-03 03:12:48.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"da78f4f4edd289d10b3a6011b13e275ae5ad7f0b","revisions":{"8a39acc86ed01140d7464910677f8a516f1edd8c":{"kind":"REWORK","_number":1,"created":"2020-12-04 15:30:55.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/1"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"subject":"[WIP] PCI Socket Affnity","message":"[WIP] PCI Socket Affnity\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffnity for a NUMA node, or no affinity at all. This is modelled on\nolder processors\u0027s layout, and has worked fine until now.\n\nOlder processors package multiple NUMA nodes into a single package,\nwhich is installed into a single physical socket. The internal\nprocessor layout makes it so that PCI devices have affinity to a\nspecific NUMA node, despite all the nodes sharing the same socket.\n\nMore recent processors, in particular from AMD, have a different\ninternal layout. This makes Nova\u0027s current NUMA node affinity model\ninsufficient. Second generation AMD EPYC processors have a central IO\nchiplet per socket, with a configurable number of NUMA nodes on top.\nPCI devices are affined to this central chiplet, as opposed to\nindividual NUMA nodes. In other words, Nova needs a way to express PCI\naffinity to a socket. This spec proposes a way to implement that.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8a39acc86ed01140d7464910677f8a516f1edd8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8a39acc86ed01140d7464910677f8a516f1edd8c"}]},"branch":"refs/heads/master"},"9008385debf6339c49068825aa4ad1bbbbe08efc":{"kind":"REWORK","_number":2,"created":"2021-01-12 21:21:58.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/2"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-01-12 21:21:48.000000000","tz":-300},"subject":"`socket` PCI NUMA-affinity Policy","message":"`socket` PCI NUMA-affinity Policy\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffinity for a NUMA node, or no affinity at all. This makes one of two\nassumptions about the underlying host NUMA topology. Either there is\nonly a single NUMA node per socket, or - for cluster on die topologies\nwith multiple nodes per socket - there are enough CPUs in each NUMA\nnode to fit reasonnably large VMs that require strict PCI NUMA\naffinity. The latter assumption is no longer true, and Nova needs a\nmore nuanced way to express PCI NUMA affinity.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9008385debf6339c49068825aa4ad1bbbbe08efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9008385debf6339c49068825aa4ad1bbbbe08efc"}]},"branch":"refs/heads/master"},"a3b0e415895819f5ea91aa6d2ad0a83805cada42":{"kind":"REWORK","_number":3,"created":"2021-01-13 19:33:30.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/3"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-01-13 19:33:10.000000000","tz":-300},"subject":"`socket` PCI NUMA-affinity Policy","message":"`socket` PCI NUMA-affinity Policy\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffinity for a NUMA node, or no affinity at all. This makes one of two\nassumptions about the underlying host NUMA topology. Either there is\nonly a single NUMA node per socket, or - for cluster on die topologies\nwith multiple nodes per socket - there are enough CPUs in each NUMA\nnode to fit reasonnably large VMs that require strict PCI NUMA\naffinity. The latter assumption is no longer true, and Nova needs a\nmore nuanced way to express PCI NUMA affinity.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a3b0e415895819f5ea91aa6d2ad0a83805cada42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a3b0e415895819f5ea91aa6d2ad0a83805cada42"}]},"branch":"refs/heads/master"},"17c48a8df696dcf78e57bb9e3d7f8268e318f3fb":{"kind":"REWORK","_number":4,"created":"2021-01-18 19:12:55.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/4"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-01-18 19:12:48.000000000","tz":-300},"subject":"`socket` PCI NUMA-affinity Policy","message":"`socket` PCI NUMA-affinity Policy\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffinity for a NUMA node, or no affinity at all. This makes one of two\nassumptions about the underlying host NUMA topology. Either there is\nonly a single NUMA node per socket, or - for cluster on die topologies\nwith multiple nodes per socket - there are enough CPUs in each NUMA\nnode to fit reasonnably large VMs that require strict PCI NUMA\naffinity. The latter assumption is no longer true, and Nova needs a\nmore nuanced way to express PCI NUMA affinity.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/17c48a8df696dcf78e57bb9e3d7f8268e318f3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/17c48a8df696dcf78e57bb9e3d7f8268e318f3fb"}]},"branch":"refs/heads/master"},"a475771be07c1e83e2435b0147a9eed879bcd95b":{"kind":"REWORK","_number":5,"created":"2021-01-19 13:11:28.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/5"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-01-19 13:11:20.000000000","tz":-300},"subject":"`socket` PCI NUMA-affinity Policy","message":"`socket` PCI NUMA-affinity Policy\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffinity for a NUMA node, or no affinity at all. This makes one of two\nassumptions about the underlying host NUMA topology. Either there is\nonly a single NUMA node per socket, or - for cluster on die topologies\nwith multiple nodes per socket - there are enough CPUs in each NUMA\nnode to fit reasonnably large VMs that require strict PCI NUMA\naffinity. The latter assumption is no longer true, and Nova needs a\nmore nuanced way to express PCI NUMA affinity.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a475771be07c1e83e2435b0147a9eed879bcd95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a475771be07c1e83e2435b0147a9eed879bcd95b"}]},"branch":"refs/heads/master"},"d973d5fd8544ebef8a43917503216171746aceba":{"kind":"REWORK","_number":6,"created":"2021-01-19 14:33:16.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/6"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-01-19 14:32:48.000000000","tz":-300},"subject":"`socket` PCI NUMA-affinity Policy","message":"`socket` PCI NUMA-affinity Policy\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffinity for a NUMA node, or no affinity at all. This makes one of two\nassumptions about the underlying host NUMA topology. Either there is\nonly a single NUMA node per socket, or - for cluster on die topologies\nwith multiple nodes per socket - there are enough CPUs in each NUMA\nnode to fit reasonnably large VMs that require strict PCI NUMA\naffinity. The latter assumption is no longer true, and Nova needs a\nmore nuanced way to express PCI NUMA affinity.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d973d5fd8544ebef8a43917503216171746aceba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d973d5fd8544ebef8a43917503216171746aceba"}]},"branch":"refs/heads/master"},"da78f4f4edd289d10b3a6011b13e275ae5ad7f0b":{"kind":"REWORK","_number":7,"created":"2021-01-19 16:05:22.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/51/765551/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/51/765551/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/51/765551/7"}}},"commit":{"parents":[{"commit":"1cd70b446a8a1a1526f4fa589234c54039ee1377","subject":"Merge \"libvirt: Store and allow the default machine type to be changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1cd70b446a8a1a1526f4fa589234c54039ee1377"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-04 15:27:56.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-01-19 16:05:17.000000000","tz":-300},"subject":"`socket` PCI NUMA-affinity Policy","message":"`socket` PCI NUMA-affinity Policy\n\nNova\u0027s current support for NUMA affinity for PCI devices is limited in\nthe kinds of affinity that it can express. Either a PCI device has\naffinity for a NUMA node, or no affinity at all. This makes one of two\nassumptions about the underlying host NUMA topology. Either there is\nonly a single NUMA node per socket, or - for cluster on die topologies\nwith multiple nodes per socket - there are enough CPUs in each NUMA\nnode to fit reasonnably large VMs that require strict PCI NUMA\naffinity. The latter assumption is no longer true, and Nova needs a\nmore nuanced way to express PCI NUMA affinity.\n\nChange-Id: Ib701e3f79185eabad389136e4411b73d04d8fa04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da78f4f4edd289d10b3a6011b13e275ae5ad7f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da78f4f4edd289d10b3a6011b13e275ae5ad7f0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
