)]}'
{"id":"openstack%2Fnova-specs~769547","triplet_id":"openstack%2Fnova-specs~master~I36d3b544afb33cdeb70dc3f2a9e067904ebfac05","project":"openstack/nova-specs","branch":"master","topic":"spec/libvirt-default-machine-type","hashtags":[],"change_id":"I36d3b544afb33cdeb70dc3f2a9e067904ebfac05","subject":"libvirt: Update instance machine type stash spec","status":"MERGED","created":"2021-01-06 14:42:05.000000000","updated":"2021-01-11 16:40:03.000000000","submitted":"2021-01-11 16:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"769547-spec/libvirt-default-machine-type","meta_rev_id":"df50f9f35bde2b06b39ca3cfac948eef5ea44a30","_number":769547,"virtual_id_number":769547,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 16:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2021-01-11 16:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-11 16:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2021-01-11 16:15:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 14:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-06 15:33:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-01-06 16:13:04.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-01-11 16:31:45.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"f9dc3c235923b8d117ffef0a809f45c090d8c0c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-06 14:42:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e260e5669a565a0b59f4b3e405ab231484059015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 14:55:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4752d841df4145835fdd2a28daeabd : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3dfc8f7b4ae434484ac1ae01f62fd08 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"03d76c26af5441cd55d5355994a4eb2c9a39b47b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-06 15:33:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"886cd1c04a1d9eb4acb986198ac7247194bd9e11","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-06 16:13:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(4 comments)\n\nDefinitely agree that sysmeta is the right place for this. Since I didn\u0027t review the original spec and missed the PTG discussion, I have a couple questions which are mostly not relevant to the actual sysmeta change. The only one I think might be a welcomed tiny rev to this is the inclusion of SHELVED in the SHUTOFF logic.","accounts_in_message":[],"_revision_number":1},{"id":"ba0f9343be77d5f060772c021a746a30d526771f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-06 17:22:22.000000000","message":"Patch Set 1: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"520d08bf2531addaff062d777d9b78123ffe4790","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-06 17:42:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da548f12fce7d3870777fc42e2e59da96856004b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-07 10:56:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a91d0ea4aa6582e75a3bc606eee824abc490758a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-07 11:45:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a3e1d1bebc10e1308c58b2465b8e4914e4da6368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 12:00:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fc68d2ae38420b9235b1ab9dd282f3 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ed375493bb4fd08257d06a298541ce : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3e9b907e4cc96fa4eabb080fa99813a53abd4795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-07 12:48:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"062b47d0733af5a60efea2fb3fdb77932b019760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 12:59:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b284659d464f14a0eb9916a9c40d98 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f33f21a5184daf9aaed56a5386ee0a : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9d7eaaef10910ad9cb518cf25b313e9c0b07fb6d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-07 14:53:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1fe05b4ad0c99e427f41c42c65061173195fd0be","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-07 15:30:03.000000000","message":"Patch Set 3: Code-Review+2\n\nStill good IMO","accounts_in_message":[],"_revision_number":3},{"id":"afbdf3eec59600cfdc2636a8d38372870ab6d7b2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-11 16:14:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nI\u0027ll fix the typos and +2+W","accounts_in_message":[],"_revision_number":3},{"id":"17324f3496af0e9fb3b96a07ceaf1e200c0dbce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-11 16:15:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"135e71f4f16cfa6f08253451036b4963e932f1d9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-11 16:15:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nJust fixed typos, Stephen was +2 before, so fast-approving.","accounts_in_message":[],"_revision_number":4},{"id":"b85c302fc8dfb5984371d9f9ef1f94b01acc719f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:28:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45400c901c424fc6b7bff83bcb8104f4 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c16ac6ef954839940f31e3875d4efc : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":4},{"id":"5a71625dbc5ec75628560b6e5cc1dd31470eccc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:28:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ecafdf52cb2f4bdee037df9e0cf9dc6b6049e135","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-01-11 16:31:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"28563840299252648a3c088d8e6e5ebd55880fb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:38:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0fa9242bbd45ddbecb8fcbd91943c9 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfb7b82cd90435caa00f5abee588e3a : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":4},{"id":"29943dfe7cf8ca6b37e9a70e1623f1fec570a9ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"df50f9f35bde2b06b39ca3cfac948eef5ea44a30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:40:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/37fa617d473c4af980d8714d003d9f6f : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0d58760cdf705f2999bf2aa341df818f20b3df82","revisions":{"8c8a26db15cb251fc0b8e980bb013f7ab0118a6a":{"kind":"REWORK","_number":1,"created":"2021-01-06 14:42:05.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/47/769547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/47/769547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/1"}}},"commit":{"parents":[{"commit":"8d78679dcc0f9e096d7fed96511e25bc18c328d4","subject":"Merge \"Update modernize-os-hypervisors-api spec, pt. 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d78679dcc0f9e096d7fed96511e25bc18c328d4"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-21 11:06:53.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-01-06 14:41:37.000000000","tz":0},"subject":"libvirt: Update instance machine type stash spec","message":"libvirt: Update instance machine type stash spec\n\nThe original spec wanted to introduce a new attribute to the Instance\nobject when discussions at the PTG had actually been around using the\nexisting system metadata store for this.\n\nSome simple style nits are also addressed in this change.\n\nChange-Id: I36d3b544afb33cdeb70dc3f2a9e067904ebfac05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8c8a26db15cb251fc0b8e980bb013f7ab0118a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8c8a26db15cb251fc0b8e980bb013f7ab0118a6a"}]},"branch":"refs/heads/master"},"6892b5ed52fbcc6edecfa0f3b6fdeb519f447cae":{"kind":"REWORK","_number":2,"created":"2021-01-07 11:45:41.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/47/769547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/47/769547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/2"}}},"commit":{"parents":[{"commit":"8d78679dcc0f9e096d7fed96511e25bc18c328d4","subject":"Merge \"Update modernize-os-hypervisors-api spec, pt. 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d78679dcc0f9e096d7fed96511e25bc18c328d4"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-21 11:06:53.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-01-07 11:42:02.000000000","tz":0},"subject":"libvirt: Update instance machine type stash spec","message":"libvirt: Update instance machine type stash spec\n\n* Replaces references to a new ``hw_machine_type`` Instance object\n  attribute with the existing ``system_metadata`` dict. This was\n  actually agreed to during the PTG but lost and replaced with the new\n  attribute sometime later.\n\n* Drops the move of the hardcoded default machine types into the\n  ``[libvirt]/hw_machine_type`` configurable as on review this is not\n  actually required to achieve the goals of this spec and would if\n  anything make things more complicated when upgrading.\n\n* Lists the new ``nova-manage`` commands and adds ``SHELVED``\n  as an allowed vm_state for the ``update_machine_type`` command.\n\n* Lists the default restrictions around updating the machine type via\n  the ``update_machine_type`` outside of a full rebuild. These include\n  between underlying machine types and from newer to older versions of the\n  same machine type.\n\n* Adds a ``--force`` flag to the ``update_machine_type`` command to\n  allow operators to workaround these restrictions.\n\n* Drops the reference to the versioned machine types from the data model\n  impact section as we are now using ``system_metadata`` that forces us\n  to use strings.\n\n* Replaces the incorrect ``SHUTOFF`` vm_state with the correct ``STOPPED``\n  state.\n\n* Replaces some incorrect uses of the term deployer and operator/admin\n  with operator.\n\nChange-Id: I36d3b544afb33cdeb70dc3f2a9e067904ebfac05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6892b5ed52fbcc6edecfa0f3b6fdeb519f447cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6892b5ed52fbcc6edecfa0f3b6fdeb519f447cae"}]},"branch":"refs/heads/master"},"ec3614260c44966db3d2849512ff6cb4598131a5":{"kind":"REWORK","_number":3,"created":"2021-01-07 12:48:19.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/47/769547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/47/769547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/3"}}},"commit":{"parents":[{"commit":"8d78679dcc0f9e096d7fed96511e25bc18c328d4","subject":"Merge \"Update modernize-os-hypervisors-api spec, pt. 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d78679dcc0f9e096d7fed96511e25bc18c328d4"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-21 11:06:53.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-01-07 12:48:12.000000000","tz":0},"subject":"libvirt: Update instance machine type stash spec","message":"libvirt: Update instance machine type stash spec\n\n* Replaces references to a new ``hw_machine_type`` Instance object\n  attribute with the existing ``system_metadata`` dict. This was\n  actually agreed to during the PTG but lost and replaced with the new\n  attribute sometime later.\n\n* Drops the move of the hardcoded default machine types into the\n  ``[libvirt]/hw_machine_type`` configurable as on review this is not\n  actually required to achieve the goals of this spec and would if\n  anything make things more complicated when upgrading.\n\n* Lists the new ``nova-manage`` commands and adds ``SHELVED``\n  as an allowed vm_state for the ``update_machine_type`` command.\n\n* Lists the default restrictions around updating the machine type via\n  the ``update_machine_type`` outside of a full rebuild. These include\n  between underlying machine types and from newer to older versions of the\n  same machine type.\n\n* Adds a ``--force`` flag to the ``update_machine_type`` command to\n  allow operators to workaround these restrictions.\n\n* Drops the reference to the versioned machine types from the data model\n  impact section as we are now using ``system_metadata`` that forces us\n  to use strings.\n\n* Replaces the incorrect ``SHUTOFF`` vm_state with the correct ``STOPPED``\n  state.\n\n* Replaces some incorrect uses of the term deployer and operator/admin\n  with operator.\n\nChange-Id: I36d3b544afb33cdeb70dc3f2a9e067904ebfac05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ec3614260c44966db3d2849512ff6cb4598131a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ec3614260c44966db3d2849512ff6cb4598131a5"}]},"branch":"refs/heads/master"},"0d58760cdf705f2999bf2aa341df818f20b3df82":{"kind":"REWORK","_number":4,"created":"2021-01-11 16:15:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/47/769547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/47/769547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/47/769547/4"}}},"commit":{"parents":[{"commit":"8d78679dcc0f9e096d7fed96511e25bc18c328d4","subject":"Merge \"Update modernize-os-hypervisors-api spec, pt. 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d78679dcc0f9e096d7fed96511e25bc18c328d4"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-21 11:06:53.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2021-01-11 16:14:54.000000000","tz":0},"subject":"libvirt: Update instance machine type stash spec","message":"libvirt: Update instance machine type stash spec\n\n* Replaces references to a new ``hw_machine_type`` Instance object\n  attribute with the existing ``system_metadata`` dict. This was\n  actually agreed to during the PTG but lost and replaced with the new\n  attribute sometime later.\n\n* Drops the move of the hardcoded default machine types into the\n  ``[libvirt]/hw_machine_type`` configurable as on review this is not\n  actually required to achieve the goals of this spec and would if\n  anything make things more complicated when upgrading.\n\n* Lists the new ``nova-manage`` commands and adds ``SHELVED``\n  as an allowed vm_state for the ``update_machine_type`` command.\n\n* Lists the default restrictions around updating the machine type via\n  the ``update_machine_type`` outside of a full rebuild. These include\n  between underlying machine types and from newer to older versions of the\n  same machine type.\n\n* Adds a ``--force`` flag to the ``update_machine_type`` command to\n  allow operators to workaround these restrictions.\n\n* Drops the reference to the versioned machine types from the data model\n  impact section as we are now using ``system_metadata`` that forces us\n  to use strings.\n\n* Replaces the incorrect ``SHUTOFF`` vm_state with the correct ``STOPPED``\n  state.\n\n* Replaces some incorrect uses of the term deployer and operator/admin\n  with operator.\n\nChange-Id: I36d3b544afb33cdeb70dc3f2a9e067904ebfac05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d58760cdf705f2999bf2aa341df818f20b3df82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d58760cdf705f2999bf2aa341df818f20b3df82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
