)]}'
{"id":"openstack%2Fnova-specs~781259","triplet_id":"openstack%2Fnova-specs~master~Ia5d77416b497ef04fb915841b8cdf300230ff00c","project":"openstack/nova-specs","branch":"master","topic":"vhost-vdpa","hashtags":[],"change_id":"Ia5d77416b497ef04fb915841b8cdf300230ff00c","subject":"Change error code when vDPA is unsupported","status":"MERGED","created":"2021-03-18 10:24:20.000000000","updated":"2021-04-14 10:56:20.000000000","submitted":"2021-04-14 10:54:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781259-vhost-vdpa","meta_rev_id":"73914a40accf3c3fbf1a572125cdd982d18955d3","_number":781259,"virtual_id_number":781259,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 10:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2021-04-14 10:56:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2021-04-14 10:30:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 10:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-20 14:10:19.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2021-03-21 18:02:44.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2021-03-23 18:47:03.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-03-30 19:30:31.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-13 15:42:10.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"4779cd0bc1cabcdf7da58fa1a5a1334ce49a2323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-18 10:24:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb79c05ab1d579429159e4ed7eb4ef9cb566bb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465619ffbf2f47ecbf9aa49db03f1128 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263b57506ad148fe960706725f3efca4 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"734f66371b5b60f7d147b7798e60123c7769919f","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-03-20 14:10:19.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM\n\nhttps://review.opendev.org/c/openstack/nova/+/780333","accounts_in_message":[],"_revision_number":1},{"id":"2f35dec18376bac3b4bb17fa3a5cbc38fad75b7c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-03-21 18:02:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"853342fd16324ef6e4d76899344a70d49e9c25cc","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-23 18:47:03.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"9a729ec0615d908f5d8218669fe228b59a6066db","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-03-30 19:30:31.000000000","message":"Patch Set 1: Code-Review+1\n\nok the 409 conflict change has been merged so this is just bring the spec in line with code.","accounts_in_message":[],"_revision_number":1},{"id":"9c4af94699aebbf556ea7f576c194dd414a100e8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-13 15:42:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5424f7bbcdb61f2ceebdbb291db0de90734ae8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:42:24.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f5467fa789d130ce2dbe35e3be211d1837654b21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-13 15:43:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e45a1c629da888a7e79a1b1d8856016d5092a14","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-13 15:44:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\njust resolved the merge conflict as  in the meantime the spec is moved to the implemented folder. fast approving","accounts_in_message":[],"_revision_number":2},{"id":"75384f5b00ffd591661a2d365f46dda12fed0498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:55:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb75e47dddb4a2bb285b21f30c0bef3 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6542ba53da8e4e0bb4881dab745abab4 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b449124f6fca88698e79d5799c66d8a77330780c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 10:30:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0e32f4b65e0cfb7058e2868cb5bcb8d7549ab11c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 10:30:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"77f4c10a7f965e177511d27cd1a2a50fe5e15e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:43:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d7f36d81e8459991819098b10e71ef : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66bf86b88f8e4ed98c05ec1101734b04 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ed85fa3cb1d6ea8ab914991a70eae9bb5fb3b922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:43:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1d1c6c9486cb13137b34b28e8864a7a99c396a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:54:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04ebdbf3fec4d528ef79dfc62d15187 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69790e415b546ab9ca43771bcf8cf00 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"8d7b5631dc9a73c0d7e5cb0dff5f6a92c50cc783","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:54:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ad417022da3afb7a89bea013a54a69257d5a3bf7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 10:56:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/491c426496fe4673aa90a86302d039b9 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43380e19d72c6d8198124f5b61c07c0b35544a3d","revisions":{"149c4fa48932e790d3783d47455eef24a6eca55f":{"kind":"REWORK","_number":1,"created":"2021-03-18 10:24:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/781259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/59/781259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/1"}}},"commit":{"parents":[{"commit":"abfb91875c7d6b1e2df45d5bc2244956bc662884","subject":"Remove policy changes from modernize-os-hypervisors-api spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/abfb91875c7d6b1e2df45d5bc2244956bc662884"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-18 09:40:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-18 09:51:18.000000000","tz":0},"subject":"Change error code when vDPA is unsupported","message":"Change error code when vDPA is unsupported\n\nWe had specified that we would return HTTP 403 (Forbidden) in the event\nthat a unsupported operation was requested on an instance with one or\nmore vDPA interfaces attached. This was chosen due to us already using\nthis code for unsupported operations with cyborg accelerators. However,\nHTTP 403 (Forbidden) isn\u0027t a great choice for this, as it suggests\nissues with authorization as opposed to a request the server simply\ncan\u0027t support. HTTP 409 (Conflict) isn\u0027t much better, but it does avoid\nconfusion with policy issues and aligns us with two other existing\nexamples: unsupported operations on instances with vTPM devices and\ninstances with SEV enabled.\n\nChange-Id: Ia5d77416b497ef04fb915841b8cdf300230ff00c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/149c4fa48932e790d3783d47455eef24a6eca55f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/149c4fa48932e790d3783d47455eef24a6eca55f"}]},"branch":"refs/heads/master"},"f7db9099e9fa9caa85c6b3f27f7bb54d5dd285e1":{"kind":"REWORK","_number":2,"created":"2021-04-13 15:43:35.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/59/781259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/59/781259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/2"}}},"commit":{"parents":[{"commit":"796534a500c817f9ba51e7184bba5218f65f8072","subject":"QoS minimum guaranteed packet rate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/796534a500c817f9ba51e7184bba5218f65f8072"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-18 09:40:40.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-13 15:42:57.000000000","tz":120},"subject":"Change error code when vDPA is unsupported","message":"Change error code when vDPA is unsupported\n\nWe had specified that we would return HTTP 403 (Forbidden) in the event\nthat a unsupported operation was requested on an instance with one or\nmore vDPA interfaces attached. This was chosen due to us already using\nthis code for unsupported operations with cyborg accelerators. However,\nHTTP 403 (Forbidden) isn\u0027t a great choice for this, as it suggests\nissues with authorization as opposed to a request the server simply\ncan\u0027t support. HTTP 409 (Conflict) isn\u0027t much better, but it does avoid\nconfusion with policy issues and aligns us with two other existing\nexamples: unsupported operations on instances with vTPM devices and\ninstances with SEV enabled.\n\nChange-Id: Ia5d77416b497ef04fb915841b8cdf300230ff00c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7db9099e9fa9caa85c6b3f27f7bb54d5dd285e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7db9099e9fa9caa85c6b3f27f7bb54d5dd285e1"}]},"branch":"refs/heads/master"},"43380e19d72c6d8198124f5b61c07c0b35544a3d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-14 10:30:26.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/59/781259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/59/781259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/59/781259/3"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-03-18 09:40:40.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-14 10:30:26.000000000","tz":0},"subject":"Change error code when vDPA is unsupported","message":"Change error code when vDPA is unsupported\n\nWe had specified that we would return HTTP 403 (Forbidden) in the event\nthat a unsupported operation was requested on an instance with one or\nmore vDPA interfaces attached. This was chosen due to us already using\nthis code for unsupported operations with cyborg accelerators. However,\nHTTP 403 (Forbidden) isn\u0027t a great choice for this, as it suggests\nissues with authorization as opposed to a request the server simply\ncan\u0027t support. HTTP 409 (Conflict) isn\u0027t much better, but it does avoid\nconfusion with policy issues and aligns us with two other existing\nexamples: unsupported operations on instances with vTPM devices and\ninstances with SEV enabled.\n\nChange-Id: Ia5d77416b497ef04fb915841b8cdf300230ff00c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43380e19d72c6d8198124f5b61c07c0b35544a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43380e19d72c6d8198124f5b61c07c0b35544a3d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
