)]}'
{"id":"openstack%2Fnova-specs~783783","triplet_id":"openstack%2Fnova-specs~master~Ie913bbe6987337570c0eadafb4847a48b3817799","project":"openstack/nova-specs","branch":"master","topic":"bp/remove-tenant-id","hashtags":[],"change_id":"Ie913bbe6987337570c0eadafb4847a48b3817799","subject":"Re-propose Remove tenant_id","status":"MERGED","created":"2021-03-30 05:54:08.000000000","updated":"2021-04-19 13:15:46.000000000","submitted":"2021-04-19 13:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783783-bp/remove-tenant-id","meta_rev_id":"a9fd8c52611b1b69c1c2562228e930d6e72a6398","_number":783783,"virtual_id_number":783783,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-19 13:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-04-19 12:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-04-13 15:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-19 13:15:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-04-19 12:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 06:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 19:27:09.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-13 15:45:20.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-04-19 00:59:15.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"fabfbf67c56d580db82f8797370c6a62c53aaf5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-03-30 05:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55c74ef4a0af733ac0cc45bb56fe49fb031bdad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 06:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bdd196985f4fcaabac56b04edca914 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18006ff07ce4615b6ba65082c6be122 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"87d4c8b95537b7a3d975cc7e8891b3428096bb67","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-03-30 19:27:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\ni belive i review this last cycle and i do not have any outstandng questions for this proposal so +1\nit improve the uniformaty of the api experience and has already been adopted in other apis so this looks good to me","accounts_in_message":[],"_revision_number":1},{"id":"9ce0b9953daba32581bdfa25e7bc01e8ef011b1c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-03-31 00:36:15.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks Sean, lgtm, updated.","accounts_in_message":[],"_revision_number":1},{"id":"26b939f401061ebcd44b7c1f9650098748586de6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-03-31 01:11:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58eded6cc3ea0bb57c0564c7a4b703d7d5ac6bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 01:25:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ecf548b9254f9cb61f34a9c4498a44 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d729ac97a5487c935bede8a2ec4105 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b904b478e7cee1a9694350f1ead3b5a24456b113","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-13 15:45:20.000000000","message":"Patch Set 2: Code-Review+2\n\nIt is a clean repropose","accounts_in_message":[],"_revision_number":2},{"id":"f6ff83f64bbc6723ba633d1851e5bdc7c5531d0d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-19 12:56:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks for reporposing this and working to improve the overall ux of the apis.","accounts_in_message":[],"_revision_number":2},{"id":"fd06481a8e9527762f6c6e40efbef98b38f06596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:56:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca3d3a49bfc3164419f168320a8004b11c0781a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 13:06:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290db4865d9741d48af0c099c6f08bfd : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb400fd0a884a42915d825710c85f39 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"386e1d719debcb287c8774dd55bf0a5d2d198bf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 13:06:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"97105bfb84f2c5c8323ee4af03f8b468afdf0f6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 13:07:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/06bf63886a724d9a841c1e75421dff8b : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a9fd8c52611b1b69c1c2562228e930d6e72a6398","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-19 13:15:46.000000000","message":"Patch Set 2: Code-Review+2\n\nate in review but lgtm, code was also close for this in Wallaby.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d6977fa2c9b63c6a095cb54d9b90dfa48533845","revisions":{"0de28a0363a775f41ad8638015afe24df923ad2a":{"kind":"REWORK","_number":1,"created":"2021-03-30 05:54:08.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/83/783783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/83/783783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/1"}}},"commit":{"parents":[{"commit":"e25ad7219f57accfc7938fb0e2732ee81b840c12","subject":"Create specs directory for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e25ad7219f57accfc7938fb0e2732ee81b840c12"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-03-30 05:53:05.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-03-30 05:54:04.000000000","tz":480},"subject":"Re-propose Remove tenant_id","message":"Re-propose Remove tenant_id\n\nThe blueprint proposes to remove the API interface that uses\n``tenant_id`` and replace it with ``project_id``, and changes\ntenant to project in all docs, policies and code.\n\nPart of blueprint remove-tenant-id\n\nChange-Id: Ie913bbe6987337570c0eadafb4847a48b3817799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0de28a0363a775f41ad8638015afe24df923ad2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0de28a0363a775f41ad8638015afe24df923ad2a"}]},"branch":"refs/heads/master"},"0d6977fa2c9b63c6a095cb54d9b90dfa48533845":{"kind":"REWORK","_number":2,"created":"2021-03-31 01:11:11.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/83/783783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/83/783783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/83/783783/2"}}},"commit":{"parents":[{"commit":"e25ad7219f57accfc7938fb0e2732ee81b840c12","subject":"Create specs directory for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e25ad7219f57accfc7938fb0e2732ee81b840c12"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-03-30 05:53:05.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-03-31 01:09:58.000000000","tz":480},"subject":"Re-propose Remove tenant_id","message":"Re-propose Remove tenant_id\n\nThe blueprint proposes to remove the API interface that uses\n``tenant_id`` and replace it with ``project_id``, and changes\ntenant to project in all docs, policies and code.\n\nPart of blueprint remove-tenant-id\n\nChange-Id: Ie913bbe6987337570c0eadafb4847a48b3817799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d6977fa2c9b63c6a095cb54d9b90dfa48533845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d6977fa2c9b63c6a095cb54d9b90dfa48533845"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
