)]}'
{"id":"openstack%2Fnova-specs~785563","triplet_id":"openstack%2Fnova-specs~master~Id858f50c3804f9d60e1e79c3c3a042d508667f5d","project":"openstack/nova-specs","branch":"master","topic":"bp/allow-migrate-pmem-data","hashtags":[],"change_id":"Id858f50c3804f9d60e1e79c3c3a042d508667f5d","subject":"Allow migrating PMEM\u0027s data","status":"MERGED","created":"2021-04-09 07:03:07.000000000","updated":"2021-05-25 10:49:41.000000000","submitted":"2021-05-25 07:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":144,"unresolved_comment_count":54,"has_review_started":true,"submission_id":"785563-bp/allow-migrate-pmem-data","meta_rev_id":"b017a4d7824ac84949ce23b38aa849b8b8013ab5","_number":785563,"virtual_id_number":785563,"owner":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-25 07:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-05-21 11:05:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-05-23 02:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2021-05-25 07:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2021-05-25 07:42:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 07:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 08:50:11.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-04-12 12:31:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-14 06:13:50.000000000","updated_by":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"reviewer":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"state":"REVIEWER"},{"updated":"2021-04-14 11:33:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-04-16 06:10:26.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2021-05-08 01:23:24.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"CC"},{"updated":"2021-05-08 01:34:14.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"}],"messages":[{"id":"814db28c1158924b022371d66079f685b2edd1bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-09 07:03:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42b506553efd34af5e03597c5154dfc7d5caeb64","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-09 07:04:26.000000000","message":"Topic set to https://blueprints.launchpad.net/nova/+spec/copying-pmem-devices-data","accounts_in_message":[],"_revision_number":1},{"id":"160372cde5f609ae3f11c3f887652cb4173f3192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 07:18:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b9fcd8049748c1a105f13d6af06f94 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5999842884445df86fd9fe8b7722467 : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"857415df96431b2d4bf36c35b87146952cfd14a7","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-09 08:47:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2528815c97621bf12a2167c789cec3068eb7385e","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-09 08:48:11.000000000","message":"Removed Code-Review+1 by Qiu Fossen \u003cqiujunting@inspur.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"b982625fb917b3d27f2eda4db87af37538ed5df4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-09 08:49:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f09776fa37067bd8c17e6bf5d7273f7b4a8bbd34","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-09 08:54:46.000000000","message":"Topic https://blueprints.launchpad.net/nova/+spec/copying-pmem-devices-data removed","accounts_in_message":[],"_revision_number":2},{"id":"2480be35bd320bf2c43f45541089e76a8025a939","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-09 08:54:57.000000000","message":"Topic set to bp/copying-pmem-devices-data","accounts_in_message":[],"_revision_number":2},{"id":"3792ba08f3575dffda9e6e1f966ea0769982b074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:56:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc43886a2d84de88861e68c217c49f0 : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7428d7d6f534bd8a0aa7b59b8aac870 : SUCCESS in 4m 21s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/wallaby/approved/copying-pmem-devices-data.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/wallaby/approved/copying-pmem-devices-data.rst","accounts_in_message":[],"_revision_number":2},{"id":"09c13fbb4ee49dbbfb292a5dd45c5d6d80ab5403","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-09 09:59:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":2},{"id":"53d1a7eb4a589a88803c3e327e058e04b6d3232b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 02:41:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"040dc1d5cac421b16bf56674ec1dd85ac0229735","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 02:43:23.000000000","message":"Topic bp/copying-pmem-devices-data removed","accounts_in_message":[],"_revision_number":3},{"id":"62f8a1025fec1865ccd82ba2332f4be1147c41df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 02:43:37.000000000","message":"Topic set to bp/allow-migrate-pmem-data","accounts_in_message":[],"_revision_number":3},{"id":"d1899f578d63af95697b5cf109bb6be796c86283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 02:53:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d3cd4f50fc4fc6a03ce45eb1d6a460 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7121a1d535410daeea9966a8ac6d61 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a4974e861bddf63b62a69717bc778a296f357259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 08:55:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"049518aed2035ef4e4f2964b79b2a4bbd0c2c30d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 08:57:28.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"255408c063084ebc508cd8ae3c078c8b2dc255fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 09:06:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401c4d10316f4e6da15054834d4cfbff : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006e305538484c5cb45df25b8129952a : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":5},{"id":"13f356458f4d6fa1f63a971728dad84a585dc1a2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-12 10:08:41.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3184d7f60e2f55d1e08f5138e48a2d2b9eb2a150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 12:12:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ee5bdf03cd6439de0b36f4dc666bbfce1443d760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 12:22:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dea58e4e6da45f4b3593df5d58b3438 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6c19f23a714ead87358a11b4215dfe : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6},{"id":"8fb321748fbb6d056119db65d580fdb7b307a173","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-12 12:31:23.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"02e049ff588c43dc87240069a91a2a4f826d7c28","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-14 02:08:35.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nYou can follow sean\u0027s suggestion, let\u0027s using optional argument to control whether copying the pmem device data.\n1. Need to add a new microverision, this should be clarified in *Proposed Change*\n2. Need to add some changes of code migrate, reisize APIs in *Rest API impact*.","accounts_in_message":[],"_revision_number":6},{"id":"f756a585fd03aeaa17509b15f11cb4a6b310021a","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2021-04-14 06:13:50.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6a113a759c2ff71992bd15c5e32e4a94a23fe15f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 11:33:12.000000000","message":"Patch Set 6: Code-Review-1\n\n(11 comments)\n\nI have also a couple of thinhs.","accounts_in_message":[],"_revision_number":6},{"id":"6e651caeb99f4c7d8f6e7e421dd1cfd3cf800ced","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-15 01:31:59.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"15df3c944e5f858134a359776a56888581dfa1e6","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2021-04-15 01:53:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"08d691136799083433c50c48b4a8e8e6258df4ca","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-15 02:04:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"139f78747b4609ef5f7079d0dc041658ecbfc8b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-15 13:49:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f0c7ea8fe8dee8923fbff8057089b4833408226f","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-15 13:51:12.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6a9ecd26d106369b67e641bc777fa9bf7ac9eedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:08:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e747d524134642b87454bf23458985 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44626e83179b4cbd882305df6d6e38a1 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"fa403cac08ad41054c15475f0fd574745c774339","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-15 18:38:26.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a6d3e011e285562e69ced5867ab96942863e35e0","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-16 06:01:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7503f235cd154ab2764fdbe4f8ef188b9e0ee5e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-16 07:30:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"57d51f8060bd72d66c9f205e7e55a14fd77f175d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-16 07:33:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8ad4989a563f94abce9b0f15a8fa9528f7046ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 07:49:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592a397891634088b6cc8224106063fb : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f940f35904b4e4c8cb9c14f654709fc : SUCCESS in 4m 08s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/xena/approved/allow-migrate-pmem-data.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/xena/approved/allow-migrate-pmem-data.rst","accounts_in_message":[],"_revision_number":9},{"id":"2bf0360c880a086fc9cb4dd635acac2278b762c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-20 02:29:25.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"46157666a167dc7a0bbccf0189260cbb0d7bfd6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:39:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bfb03d8040468aaa03765a6a3e1526 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8da32e241ec43a2a6e7af340c36c5da : SUCCESS in 4m 24s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/xena/approved/allow-migrate-pmem-data.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/xena/approved/allow-migrate-pmem-data.rst","accounts_in_message":[],"_revision_number":10},{"id":"f596581f2e57050cc798acf3469156c760cc23e1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-04 17:07:18.000000000","message":"Patch Set 10: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"88e19ed9360036c2c083a417686179e94c66caf9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-04 17:43:32.000000000","message":"Patch Set 10: Code-Review-1\n\n(8 comments)\n\n-1 is mainly for gibi\u0027s comments.\n\nassuming we introduce the copy_pmem_devices parameter then the upgrade impact section and performance impact section i believe are correct.\n\ni agree with gibi however in that i don\u0027t think the work flow section is correct\nand I\u0027m unclear as to\nthe cinder interaction scenarios. see comments in-line.","accounts_in_message":[],"_revision_number":10},{"id":"04c699fda752aeae86fc4913eec6c02d062738b0","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-05 09:25:19.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4d5ce61842b2e25a91c73097c5c2c90825c48566","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-05 13:37:25.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0e0c5cea9e278367ea8c7020f22c9bd779c11053","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-06 01:43:50.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ba6310a72fb8092557b723e639f22123042b89f8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-06 08:04:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1d84f1a1b176fa158aa22d09a68fe05c1496e53a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-06 09:45:44.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b64d050de1b7fbee832d68a8a525d5baa3535bc5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-06 09:53:59.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"93c70cc6f7b9f46c618d8de16455dddce430dbe8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-07 12:54:13.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9f1f0960ecc08cadf12f84165c5606470c86e3f8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-07 18:43:43.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6480adc10ac4bcd10da0e93c468c01c605d992ec","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2021-05-08 01:23:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3e550da1030c1a94ba53b900de5b4c2e597f69d5","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2021-05-08 01:34:14.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ba3cc50a3607c9a4d83ca0eea199745cce22af42","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-10 07:52:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b7759609be4d960befd3eaa0ef419db73e6a125f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-10 09:42:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e139b53e8ebc5baad441edeb2e2ffae8fb004cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-10 09:51:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"edf8910874d2c5d0edc6fd17cbd3661760b14ccc","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-10 09:52:17.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c245c09e847c4de629796faa09884f512d5ea841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:06:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf71bd850434c66814fdf54226867a7 : FAILURE in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e908da619a04f0687ac684c2a16ac82 : FAILURE in 4m 33s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/xena/approved/allow-migrate-pmem-data.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/xena/approved/allow-migrate-pmem-data.rst","accounts_in_message":[],"_revision_number":11},{"id":"1012b32d12cf3eaf2c95dfde39cfee799609b4de","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-10 12:52:43.000000000","message":"Patch Set 11: Code-Review-1\n\n(16 comments)\n\nThis i think is much closer to being ready.\ni have some wording changes inline but overall i think this is going in the right direction.","accounts_in_message":[],"_revision_number":11},{"id":"f4882bdae851694a779b03893e79d2e93ae6c725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-11 02:52:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5d89bb7d886a4b228f6a5ab6d3f9424c76381195","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-11 02:52:56.000000000","message":"Patch Set 11:\n\n(15 comments)","accounts_in_message":[],"_revision_number":11},{"id":"43a21ae091d07528a17e35720f4376529ecc289d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 03:02:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30acafe3db984664bf4d4712e8c5fa2d : FAILURE in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1145673d0ddd4081be279369fbe94f56 : FAILURE in 3m 18s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/xena/approved/allow-migrate-pmem-data.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/xena/approved/allow-migrate-pmem-data.rst","accounts_in_message":[],"_revision_number":12},{"id":"4f8e2533cf077effe8a3366942f9544eabeadd48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-14 03:09:51.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"769c35018f512a788e9145b16d974a004ed577a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 03:23:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20dc8e0d99d6446c8a7aec29de61d781 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba72890a0784b35a1eacb14f4deac7e : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":13},{"id":"96b8a5ff7440a4f039378b1eb69795885f3c6c81","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2021-05-18 11:00:04.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"68428fda30a00980825e09ca8cdb55b0f173d2ff","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2021-05-18 11:08:59.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6b63f5e21e8472849c151f1fd24988144774a77e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-18 13:21:48.000000000","message":"Patch Set 13: Code-Review-1\n\n(5 comments)\n\nthis is a rather soft -1 i think we are more or less arrived at a workable desing but there a few issue in line that i would like to see adress. once adress i think im more or less +2 on this but i would like alex or others to be the +w on this spec.","accounts_in_message":[],"_revision_number":13},{"id":"fdb2e30f07582d0a01e30de5b8c2657aa4a7bd5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-21 09:19:25.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e8973f31efe8cad2db68686957a57e73f6f84f2d","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-05-21 09:19:39.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c1f3326ee4a799514503b0d0e4642de86901eb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:35:03.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56e7f56ec7c47cfb368b1af77eece29 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210a875dd67d4efc91fdfb178806b78d : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":14},{"id":"837c5266304f0f89323a6c1aa73cad007d09d5d6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-21 11:05:46.000000000","message":"Patch Set 14: Code-Review+2\n\n(6 comments)\n\nwhile i have some nit i don\u0027t think any of them rise to the level of needing a new revision.\nthat said if you do revise the spec it would be nice to address them.\n\ni think the spec is relatively clear now.","accounts_in_message":[],"_revision_number":14},{"id":"64d13f6f407a33ed7b406a895bfc24cea37741e3","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-23 02:38:26.000000000","message":"Patch Set 14: Code-Review+1\n\nNice job.","accounts_in_message":[],"_revision_number":14},{"id":"7d90948277e0c79cc43ce3410d18c1632646c858","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2021-05-25 07:42:50.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)\n\ngood to have another patch fix those nits.","accounts_in_message":[],"_revision_number":14},{"id":"bf39534568da1321338189767494d02dab9e6576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 07:43:07.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"2fe0eeaf0e8697fa801fbd4b90666bc0f97621f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 07:58:19.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f17ed5f7ff34ce7829f0266c6c09837 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e862bee0bf034b8db2295914feb85fb6 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":14},{"id":"9b7cb51ea9b0a558606066a830980bcb5674669d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 07:58:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"0b5a0d1379126cdcaa976767f459ab04a3a9ced6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 08:00:23.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/75e008e4c30749c594b7bab40c570970 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":14},{"id":"b017a4d7824ac84949ce23b38aa849b8b8013ab5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-25 10:49:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"8b87224de62bc4d006d2fbdb90b86465360a2208","revisions":{"b0e31f7728a40b28c41ac767eb8969cad0a0e22e":{"kind":"REWORK","_number":1,"created":"2021-04-09 07:03:07.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/1"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:19.000000000","tz":-240},"subject":"Repropose copying pmem devices data spec","message":"Repropose copying pmem devices data spec\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b0e31f7728a40b28c41ac767eb8969cad0a0e22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b0e31f7728a40b28c41ac767eb8969cad0a0e22e"}]},"branch":"refs/heads/master"},"bd5791cf53a4122f63fc29bce0187da2c56117dd":{"kind":"REWORK","_number":2,"created":"2021-04-09 08:49:22.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/2"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"Qiu Fossen","email":"qiujunting@inspur.com","date":"2021-04-09 08:46:56.000000000","tz":0},"subject":"Repropose copying pmem devices data spec","message":"Repropose copying pmem devices data spec\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bd5791cf53a4122f63fc29bce0187da2c56117dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bd5791cf53a4122f63fc29bce0187da2c56117dd"}]},"branch":"refs/heads/master"},"6fcd05a19144a185c5d023acc61f4dfaa500e8d0":{"kind":"REWORK","_number":3,"created":"2021-04-12 02:41:17.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/3"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 02:39:14.000000000","tz":-240},"subject":"Repropose copying pmem devices data spec","message":"Repropose copying pmem devices data spec\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6fcd05a19144a185c5d023acc61f4dfaa500e8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6fcd05a19144a185c5d023acc61f4dfaa500e8d0"}]},"branch":"refs/heads/master"},"bf1380846b4ffa391ba5b93760c0f1b4c908dee4":{"kind":"REWORK","_number":4,"created":"2021-04-12 08:55:01.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/4"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 08:51:17.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM\ndevices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bf1380846b4ffa391ba5b93760c0f1b4c908dee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bf1380846b4ffa391ba5b93760c0f1b4c908dee4"}]},"branch":"refs/heads/master"},"ef43ea4c411819d5d70dffdcb754380a46fab73d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-04-12 08:57:28.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/5"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 08:56:41.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ef43ea4c411819d5d70dffdcb754380a46fab73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ef43ea4c411819d5d70dffdcb754380a46fab73d"}]},"branch":"refs/heads/master"},"edd3db036f0f6b4e5fde12640f5da83a5bcf0cae":{"kind":"REWORK","_number":6,"created":"2021-04-12 12:12:09.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/6"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 12:11:07.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/edd3db036f0f6b4e5fde12640f5da83a5bcf0cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/edd3db036f0f6b4e5fde12640f5da83a5bcf0cae"}]},"branch":"refs/heads/master"},"530fcf4bff05eb857f6b5e87d4254566aebd2d9d":{"kind":"REWORK","_number":7,"created":"2021-04-15 13:49:09.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/7"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-15 13:47:35.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/530fcf4bff05eb857f6b5e87d4254566aebd2d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/530fcf4bff05eb857f6b5e87d4254566aebd2d9d"}]},"branch":"refs/heads/master"},"79b114bfbf8d686016f4e0839374b6e1906d64fb":{"kind":"REWORK","_number":8,"created":"2021-04-16 07:30:54.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/8"}}},"commit":{"parents":[{"commit":"43380e19d72c6d8198124f5b61c07c0b35544a3d","subject":"Change error code when vDPA is unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43380e19d72c6d8198124f5b61c07c0b35544a3d"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-16 07:30:06.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79b114bfbf8d686016f4e0839374b6e1906d64fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79b114bfbf8d686016f4e0839374b6e1906d64fb"}]},"branch":"refs/heads/master"},"0a7e7c8a2ef33626d72d07283058bcc854629bba":{"kind":"REWORK","_number":9,"created":"2021-04-16 07:33:33.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/9"}}},"commit":{"parents":[{"commit":"43380e19d72c6d8198124f5b61c07c0b35544a3d","subject":"Change error code when vDPA is unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43380e19d72c6d8198124f5b61c07c0b35544a3d"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-16 07:33:04.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0a7e7c8a2ef33626d72d07283058bcc854629bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0a7e7c8a2ef33626d72d07283058bcc854629bba"}]},"branch":"refs/heads/master"},"6f53c4ada8ca5a784342a1074d50155325a09e6f":{"kind":"REWORK","_number":10,"created":"2021-04-20 02:29:25.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/10"}}},"commit":{"parents":[{"commit":"846e26f61ece052d512bbce3596bbad5b4c0e4e4","subject":"Merge \"Re-propose Remove tenant_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/846e26f61ece052d512bbce3596bbad5b4c0e4e4"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-20 02:29:02.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6f53c4ada8ca5a784342a1074d50155325a09e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6f53c4ada8ca5a784342a1074d50155325a09e6f"}]},"branch":"refs/heads/master"},"e5a8e8eb9b92b9eda1118e069cbecadac54b4bec":{"kind":"REWORK","_number":11,"created":"2021-05-10 09:51:21.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/11"}}},"commit":{"parents":[{"commit":"248cbfaba4cfbc6696786599ee9d7638f84fa386","subject":"Merge \"Adds spec for supporting @ and . in keypair name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/248cbfaba4cfbc6696786599ee9d7638f84fa386"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-05-10 09:50:07.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5a8e8eb9b92b9eda1118e069cbecadac54b4bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5a8e8eb9b92b9eda1118e069cbecadac54b4bec"}]},"branch":"refs/heads/master"},"665b5f83cdfb671ef531f6c9b1f7a66c25bf4f61":{"kind":"REWORK","_number":12,"created":"2021-05-11 02:52:47.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/12"}}},"commit":{"parents":[{"commit":"248cbfaba4cfbc6696786599ee9d7638f84fa386","subject":"Merge \"Adds spec for supporting @ and . in keypair name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/248cbfaba4cfbc6696786599ee9d7638f84fa386"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-05-11 02:52:12.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/665b5f83cdfb671ef531f6c9b1f7a66c25bf4f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/665b5f83cdfb671ef531f6c9b1f7a66c25bf4f61"}]},"branch":"refs/heads/master"},"a0b589f03eac0997bbb78db0707d2058ea1aa054":{"kind":"REWORK","_number":13,"created":"2021-05-14 03:09:51.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/13"}}},"commit":{"parents":[{"commit":"248cbfaba4cfbc6696786599ee9d7638f84fa386","subject":"Merge \"Adds spec for supporting @ and . in keypair name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/248cbfaba4cfbc6696786599ee9d7638f84fa386"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-05-14 03:08:26.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a0b589f03eac0997bbb78db0707d2058ea1aa054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a0b589f03eac0997bbb78db0707d2058ea1aa054"}]},"branch":"refs/heads/master"},"8b87224de62bc4d006d2fbdb90b86465360a2208":{"kind":"REWORK","_number":14,"created":"2021-05-21 09:19:25.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/63/785563/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/63/785563/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/63/785563/14"}}},"commit":{"parents":[{"commit":"248cbfaba4cfbc6696786599ee9d7638f84fa386","subject":"Merge \"Adds spec for supporting @ and . in keypair name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/248cbfaba4cfbc6696786599ee9d7638f84fa386"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-09 06:58:12.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-05-21 09:18:41.000000000","tz":-240},"subject":"Allow migrating PMEM\u0027s data","message":"Allow migrating PMEM\u0027s data\n\nThis spec will be make the instance support code migrate if it allocate PMEM devices.\n\nChange-Id: Id858f50c3804f9d60e1e79c3c3a042d508667f5d\nImplements: blueprint allow-migrate-pmem-data\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b87224de62bc4d006d2fbdb90b86465360a2208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b87224de62bc4d006d2fbdb90b86465360a2208"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
