)]}'
{"id":"openstack%2Fnova-specs~785674","triplet_id":"openstack%2Fnova-specs~master~I1333842fbac00574887fd1bfe699c8e4185fd334","project":"openstack/nova-specs","branch":"master","topic":"contrib-allow-special-characters-in-keypair-name","hashtags":[],"change_id":"I1333842fbac00574887fd1bfe699c8e4185fd334","subject":"Adds spec for supporting @ and . in keypair name","status":"MERGED","created":"2021-04-09 15:48:53.000000000","updated":"2021-05-04 16:56:54.000000000","submitted":"2021-05-04 16:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"785674-contrib-allow-special-characters-in-keypair-name","meta_rev_id":"fc6a50bea3fc003a2e6dd8838bc1fda02e8ad070","_number":785674,"virtual_id_number":785674,"owner":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 16:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-05-04 16:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-05-04 15:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-05-04 16:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 15:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 17:30:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-14 11:36:19.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"c70543c170b6ddafa4910860a8e9377852119224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"date":"2021-04-09 15:48:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63df419465a454292372287eb299c12e3822be18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 15:59:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9218eb605f3c46e5bddf9bb3a2784e12 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd02bc6df5a4905a35da374266cf126 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"fb55ac6daa6605be813b1dea95acb0dedfb01a8b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-09 17:30:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)\n\n-1 is mainly for the links and move some infor out of work items to the problem description.\nover all i think this makes sense.\nsee comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"d90a7168371a7039b2bb0b1504244b1392e48b99","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 11:36:19.000000000","message":"Patch Set 1: Code-Review+1\n\nThe direction looks good to me. Please fix Sean\u0027s comments and I will upgrade my vote to +2","accounts_in_message":[],"_revision_number":1},{"id":"47a6908bd6fe1806063766de68fc98a5aeb1672b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"date":"2021-04-20 09:33:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94438e735b31c52db86d361cf29c373ab3d5fbd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"date":"2021-04-20 09:36:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f1184af8e0576f67434c78a791a9eff2d9ca6065","author":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"date":"2021-04-20 09:40:38.000000000","message":"Patch Set 3:\n\n(8 comments)\n\nThank you for your inputs.\nI resolved comments when I believe I handled them, but unsure of what the habits are about this. Don\u0027t hesitate to reopen if needed!","accounts_in_message":[],"_revision_number":3},{"id":"ea4dcf3fb8bc6a6156430f4ef45dea1ce6b75d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 09:47:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a934f96fef746b59944776d824ff345 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e397de27e1b4d86b5f983d169d23981 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"2b4c7d00bd718fd1d5a09f4e4308785fa5540736","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-04 15:37:52.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nThanks. It looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"28364a4f32c3462c23a642b11ff449812dd50bde","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-04 16:46:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis is a trivial but useful enhancement.\nover all im +2 on this i don\u0027t think there are any open questions.\nwe had discussed this previously as a possible specless blueprint if it\nwas not for the need to have a micro version so given gibi\u0027s assent ill also +w","accounts_in_message":[],"_revision_number":3},{"id":"21cab77ad3c591d360e0461fdb31fe943305f46b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 16:46:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3d07cbc585507e2c2755f8d320fc19f060486d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 16:54:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/346427ce129d44b1a1bebc40d16183a3 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db735acbbe074fb49332286d1f704ad5 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"5723591e3af5c8deb7afd7b9aa9dd763209710d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 16:54:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fc6a50bea3fc003a2e6dd8838bc1fda02e8ad070","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 16:56:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e3593281c65b46b1b626e32ece6b182e : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16fbb837d5f35ecfa9647c49ee94e6e0ee3bf616","revisions":{"a4bdcd58af6808e283f2b6170152a4e01ee1a849":{"kind":"REWORK","_number":1,"created":"2021-04-09 15:48:53.000000000","uploader":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"ref":"refs/changes/74/785674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/74/785674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/1"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","date":"2021-04-09 10:07:09.000000000","tz":120},"committer":{"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","date":"2021-04-09 15:48:13.000000000","tz":120},"subject":"Adds spec for supporting @ and . in keypair name","message":"Adds spec for supporting @ and . in keypair name\n\nhttps://blueprints.launchpad.net/nova/+spec/allow-special-characters-in-keypair-name\n\nAllowing the \u0027@\u0027 and \u0027.\u0027 (dot) characters in keypair names will improve\nend user experience, as some already try to input names in format user.name@host.\nThe default comment generated by ssh-keygen that users also try to use,\nis also in this format.\n\nAPIImpact\n\nImplements: blueprint allow-special-characters-in-keypair-name\nChange-Id: I1333842fbac00574887fd1bfe699c8e4185fd334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a4bdcd58af6808e283f2b6170152a4e01ee1a849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a4bdcd58af6808e283f2b6170152a4e01ee1a849"}]},"branch":"refs/heads/master"},"b3c3d77b1961e9f05c5ea732ad7384c9d7003450":{"kind":"REWORK","_number":2,"created":"2021-04-20 09:33:37.000000000","uploader":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"ref":"refs/changes/74/785674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/74/785674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/2"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","date":"2021-04-09 10:07:09.000000000","tz":120},"committer":{"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","date":"2021-04-20 09:33:15.000000000","tz":120},"subject":"Adds spec for supporting @ and . in keypair name","message":"Adds spec for supporting @ and . in keypair name\n\nhttps://blueprints.launchpad.net/nova/+spec/allow-special-characters-in-keypair-name\n\nAllowing the \u0027@\u0027 and \u0027.\u0027 (dot) characters in keypair names will improve\nend user experience, as some already try to input names in format user.name@host.\nThe default comment generated by ssh-keygen that users also try to use,\nis also in this format.\n\nAPIImpact\n\nImplements: blueprint allow-special-characters-in-keypair-name\nChange-Id: I1333842fbac00574887fd1bfe699c8e4185fd334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3c3d77b1961e9f05c5ea732ad7384c9d7003450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3c3d77b1961e9f05c5ea732ad7384c9d7003450"}]},"branch":"refs/heads/master"},"16fbb837d5f35ecfa9647c49ee94e6e0ee3bf616":{"kind":"REWORK","_number":3,"created":"2021-04-20 09:36:11.000000000","uploader":{"_account_id":32637,"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","username":"Nautik"},"ref":"refs/changes/74/785674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/74/785674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/74/785674/3"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","date":"2021-04-09 10:07:09.000000000","tz":120},"committer":{"name":"Nicolas Parquet","email":"nicolas.parquet@gandi.net","date":"2021-04-20 09:36:01.000000000","tz":120},"subject":"Adds spec for supporting @ and . in keypair name","message":"Adds spec for supporting @ and . in keypair name\n\nhttps://blueprints.launchpad.net/nova/+spec/allow-special-characters-in-keypair-name\n\nAllowing the \u0027@\u0027 and \u0027.\u0027 (dot) characters in keypair names will improve\nend user experience, as some already try to input names in format user.name@host.\nThe default comment generated by ssh-keygen that users also try to use,\nis also in this format.\n\nAPIImpact\n\nImplements: blueprint allow-special-characters-in-keypair-name\nChange-Id: I1333842fbac00574887fd1bfe699c8e4185fd334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/16fbb837d5f35ecfa9647c49ee94e6e0ee3bf616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/16fbb837d5f35ecfa9647c49ee94e6e0ee3bf616"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
