)]}'
{"id":"openstack%2Fnova-specs~785860","triplet_id":"openstack%2Fnova-specs~master~I3eee4e776e1a812b2b6112aa4bc463575efeb366","project":"openstack/nova-specs","branch":"master","topic":"768853","hashtags":[],"change_id":"I3eee4e776e1a812b2b6112aa4bc463575efeb366","subject":"Support live migrate vtpm server","status":"ABANDONED","created":"2021-04-12 11:59:43.000000000","updated":"2021-04-15 02:14:21.000000000","total_comment_count":6,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"84b92acb5b9b088ab681b9b2691f8c9d96568c8e","_number":785860,"virtual_id_number":785860,"owner":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-15 02:10:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2021-04-15 02:14:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-12 12:01:20.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-04-12 12:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 12:19:36.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-14 12:44:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"c9256d212629071a0bac56ca15b36f3b4f76772a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 11:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"834b5cb8d5932be38f54efa01fbdf2ff085eb856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 12:09:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e1e9b91aa34fd28d2add02b8e563c6 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512a9d753726499384cbf87e84429af0 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f04c38838406fbcd0922e23eb0bbbac6fe54dc20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-12 12:17:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd06af6b0e098aa856ec07f8a4362e24fb414e64","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-12 12:19:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b74f1ee9448a60780c4aaf3964d9e29d326dcbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 12:26:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a93a317d8b77454fb091e5b97d9f400a : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5071531d0be54e2da0c7dee57699a1d0 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e7628cbcd869657b1a64a1fcd2f905bfd8d64488","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-14 12:44:07.000000000","message":"Patch Set 2: Code-Review-2\n\n(5 comments)\n\n-2 as this should be a bug report","accounts_in_message":[],"_revision_number":2},{"id":"828c7b1cb3c49a0f5a487f70a290f0adb9519c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-15 01:57:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e29255e4286160283a2e1992b4f4fea9032a3bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 02:10:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c35d9a7ba54f4eae4ce826dc8ff294 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078213293214457688f14aeffacf69ed : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"c69b9aafe6a76a988fda12c4d439e4b77c3b0c99","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-04-15 02:14:21.000000000","message":"Abandoned\n\nReferench: https://bugs.launchpad.net/nova/+bug/1924202","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aa7aaad509d6b106066c04af928e856213f04ae7","revisions":{"0ba9d21afb1118ae996b11b24532e6deedbbe834":{"kind":"REWORK","_number":1,"created":"2021-04-12 11:59:43.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/60/785860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/60/785860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/1"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 11:56:23.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 11:56:32.000000000","tz":-240},"subject":"Support live migrate vtpm server","message":"Support live migrate vtpm server\n\nThis spec proposes an method for live migrating instance with vTPM device.\n\nChange-Id: I3eee4e776e1a812b2b6112aa4bc463575efeb366\nblueprint: support-live-migrate-vtpm-server\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ba9d21afb1118ae996b11b24532e6deedbbe834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ba9d21afb1118ae996b11b24532e6deedbbe834"}]},"branch":"refs/heads/master"},"a3dd7ded0c5b82c7ecc822f317a925169405af69":{"kind":"REWORK","_number":2,"created":"2021-04-12 12:17:08.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/60/785860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/60/785860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/2"}}},"commit":{"parents":[{"commit":"f7394b308818e6b921cf28beff976eac635a301e","subject":"Move the implemented specs for the wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7394b308818e6b921cf28beff976eac635a301e"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 11:56:23.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 12:16:38.000000000","tz":-240},"subject":"Support live migrate vtpm server","message":"Support live migrate vtpm server\n\nThis spec proposes an method for live migrating instance with vTPM device.\n\nChange-Id: I3eee4e776e1a812b2b6112aa4bc463575efeb366\nblueprint: support-live-migrate-vtpm-server\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a3dd7ded0c5b82c7ecc822f317a925169405af69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a3dd7ded0c5b82c7ecc822f317a925169405af69"}]},"branch":"refs/heads/master"},"aa7aaad509d6b106066c04af928e856213f04ae7":{"kind":"REWORK","_number":3,"created":"2021-04-15 01:57:43.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/60/785860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/60/785860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/60/785860/3"}}},"commit":{"parents":[{"commit":"43380e19d72c6d8198124f5b61c07c0b35544a3d","subject":"Change error code when vDPA is unsupported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/43380e19d72c6d8198124f5b61c07c0b35544a3d"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-12 11:56:23.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2021-04-15 01:54:47.000000000","tz":-240},"subject":"Support live migrate vtpm server","message":"Support live migrate vtpm server\n\nThis spec proposes an method for live migrating instance with vTPM device.\n\nChange-Id: I3eee4e776e1a812b2b6112aa4bc463575efeb366\nblueprint: support-live-migrate-vtpm-server\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aa7aaad509d6b106066c04af928e856213f04ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aa7aaad509d6b106066c04af928e856213f04ae7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
