)]}'
{"id":"openstack%2Fnova-specs~788243","triplet_id":"openstack%2Fnova-specs~master~I5510ebc584842aac5c999fbcce0ff0359016a1f9","project":"openstack/nova-specs","branch":"master","topic":"support-re-parenting-providers","hashtags":[],"change_id":"I5510ebc584842aac5c999fbcce0ff0359016a1f9","subject":"Allow provider re-parenting in placement","status":"MERGED","created":"2021-04-27 10:54:18.000000000","updated":"2021-06-02 18:03:00.000000000","submitted":"2021-06-02 17:58:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":14,"has_review_started":true,"submission_id":"788243-support-re-parenting-providers","meta_rev_id":"cd4cf111a76515e13130580ee216d7b294207e8b","_number":788243,"virtual_id_number":788243,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 17:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2021-06-02 17:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2021-06-02 12:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2021-06-02 08:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2021-06-02 17:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 10:56:28.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-27 10:56:28.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2021-04-27 10:56:28.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-04-27 11:13:02.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2021-04-27 11:13:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 15:58:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2021-05-26 14:46:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"d6b8961c61a976d1201bb0ef75ff1f3350cb0f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-27 10:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0a04fa50e13d1c6eb58e1fc8aed6983f1b87359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:13:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd944916b13c42a7b6f1047e4f2c9472 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5882ee7ad134a979bf5ad90c009e75c : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bb0501b6508aec85ab37cc2035abe446538a4c21","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-04-29 21:24:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nLooks reasonable to me, +1 for now while other reviewers take a look.","accounts_in_message":[],"_revision_number":1},{"id":"7e8c03197eab3681a4949c81b1954ddf7eaca24c","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-01 01:33:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"173a8dbeb9b5884fae6c5a169f69eba01ecafe91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-04 14:54:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0f2b0f87c6734805cc3f33c429281ffe6e37a4eb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-04 14:55:29.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nThanks for the reviews I fixed the nits and responded to the comments.","accounts_in_message":[],"_revision_number":2},{"id":"20aeab5907d7922139b58407a4eb279f0503c9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:05:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e09e6461e564e89abdfc42c28f4277a : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15282f992c1e4ccbb01d05b0a0b89968 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d45a5beb81e7491bdcf54ede2effd4c043b0d001","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-05-11 16:32:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)\n\nSoft +1 as I\u0027m not really happy with using this kind of hacky query parameter. You really want to do things, whatever they are.","accounts_in_message":[],"_revision_number":2},{"id":"995ed9f0b2f0f8987bac78cf6a2729390f0be10b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-12 10:05:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"548ada984abb4cb355cea71a42abacc552f6f501","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-12 10:06:58.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bddc49407ba0f29b0d2dd984a352e75367fc56c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:16:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7a1afce8a4431faa3e882e1c7d5511 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0193d0e9254109a73836181de25ebb : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"aa756eddaf057901dc2c5fb13bdd89cd86a44243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-13 15:00:11.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"02175dd99f6cae213b323da31449f225b9284afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:10:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb45f5e4e4574712896122acaabeb323 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815fa841cfa64a9c90426e37b356ff57 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"62fb3010120185c8152105d31739986f7c2c70ec","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-05-25 12:58:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)\n\nI\u0027m not super happy with this QS param and I\u0027m trying to think about the better API.\nThat being said, I don\u0027t wanna hold on this hence the soft +1.","accounts_in_message":[],"_revision_number":4},{"id":"724d66d8d7c59e99874a02f8e6ad99ef3fb8166e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-25 13:32:25.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nnot a full review but ill take a look at this again later today.","accounts_in_message":[],"_revision_number":4},{"id":"aff77921371b9ebc50421aba3ad5ed9419532d59","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-05-25 22:05:29.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"28c19d49f96aeeacdf4702361e9d52598569cf52","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-26 13:19:43.000000000","message":"Patch Set 4: Code-Review-1\n\nI read back the conversation. What struck me is that Tetsuro[1], Sylvain[2], Sean[3], Melanie[4] all said in some way that we don\u0027t need the are-you-sure check for re-parenting. \n\nThen we went debating how to add such a check and honestly I think we failed to find a solution that looks good to all of us. \n\nSo if all of the reviewers so far said we don\u0027t need the check in the first place then I think we just simply changed our standpoint since the PTG about such check. Which is fine. I will go and make the Do nothing as the primary alternative to see if we agree on that :)\n\nCheers,\ngibi\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/788243/2/specs/xena/approved/placement-allow-provider-re-parenting.rst#72\n[2] https://review.opendev.org/c/openstack/nova-specs/+/788243/2/specs/xena/approved/placement-allow-provider-re-parenting.rst#72\n[3] https://review.opendev.org/c/openstack/nova-specs/+/788243/4/specs/xena/approved/placement-allow-provider-re-parenting.rst#122\n[4] https://review.opendev.org/c/openstack/nova-specs/+/788243/4/specs/xena/approved/placement-allow-provider-re-parenting.rst#118","accounts_in_message":[],"_revision_number":4},{"id":"d966c3748484ab6faf7849dda479cd285df1403d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-26 13:38:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"563de531840c3162e4755baaaca8fafc4c512e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:48:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cb47fee8634878862d6c4dc59ee737 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930372716f6d43b9971039dcd29fc2b2 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":5},{"id":"b66bfbecc3a9e7e497e2115f14c3b9068f29956c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-26 14:46:40.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nWell detailed, well thought out spec. Looks like you\u0027ve got lots of feedback already from people and CI is failing so no +2 from me, but I can add that once others have had a chance to revisit and you\u0027ve addressed the CI issues","accounts_in_message":[],"_revision_number":5},{"id":"ae78fe21151745ebfd6dcc3c8813af917c1bea2c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-05-27 21:48:43.000000000","message":"Patch Set 5: Code-Review+1\n\nUpdates to the content LGTM. +1 for now while CI is red and other reviewers revisit.","accounts_in_message":[],"_revision_number":5},{"id":"1137b723b8fbd1aadac7d1bdeb888ddef91471cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-28 11:43:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c456bcd62f45c797f4cd38f6e30ca206262c4e36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-28 11:45:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"054a537d7a9de3df8953da9b4fb719045b0fe55b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-28 11:46:16.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fb403d42522d469ababe98dd9dae454071b26e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 12:21:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b8c92b1d1043b99089465ae3c05183 : TIMED_OUT in 34m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f06e534bd7477da6be9c203ef131ca : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":7},{"id":"74b9aca2f414ed15da84e72fe568e1872181f027","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-31 16:54:46.000000000","message":"Patch Set 7:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":7},{"id":"62dfaf2b7c7b83babb02b59403880f1c3c3dfefb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 17:05:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1901d8296926444a8c19c523c789c8ca : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b2bb5b3179410ba55e0cd742a369b7 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":7},{"id":"04ac2a81e5c84bcf48031021f7dfef90d0020994","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-06-02 08:28:39.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)\n\nWe could maybe discuss on some client support, but that could be a FUP.","accounts_in_message":[],"_revision_number":7},{"id":"5c3af073a18b541fb2d8b78079f18d7cda065f6b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-02 12:45:59.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)\n\nthis looks good to me but ill hold off on +w to give melanie  and other a chance to comment.","accounts_in_message":[],"_revision_number":7},{"id":"74e30a6ce4fa3e5098ad52c69881b65ee562c76c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-06-02 17:03:28.000000000","message":"Patch Set 7: Code-Review+2\n\nLooks good, thanks!","accounts_in_message":[],"_revision_number":7},{"id":"8c1aa97694a19a353e027dce1b07fce14f132796","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-06-02 17:44:11.000000000","message":"Patch Set 7: Workflow+1\n\nGot confirmation from stephenfin that he\u0027s good with this too, so +W\u0027ing.","accounts_in_message":[],"_revision_number":7},{"id":"890a60fc7c08a8eec1423128abb4fc6c7f73d920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 17:44:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cc3ec06668c182bb7d613a0ecc3384643ef2dd05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 17:58:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339f73b8b905469a9dcaa176795eae97 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866bfa5a673e48708cd27f767619ca56 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":7},{"id":"b651ac3949ed609e19028b46e8d67291d0eddcfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 17:58:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"cd4cf111a76515e13130580ee216d7b294207e8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 18:03:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/7ffc341c242f465ba6de29f662b105e7 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"15cce573ff839419f4cc8f1dd23a771070b109e0","revisions":{"ba03787689a2e8ef61a4fe66aa27d6aa58fd232a":{"kind":"REWORK","_number":1,"created":"2021-04-27 10:54:18.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/1"}}},"commit":{"parents":[{"commit":"8411c164db8884dbb778d4defdb92369867550c0","subject":"Merge \"trivial: Fix spelling, formatting of vDPA spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8411c164db8884dbb778d4defdb92369867550c0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ba03787689a2e8ef61a4fe66aa27d6aa58fd232a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ba03787689a2e8ef61a4fe66aa27d6aa58fd232a"}]},"branch":"refs/heads/master"},"ba04008b3499dd8877167b30352f90e210b11e78":{"kind":"REWORK","_number":2,"created":"2021-05-04 14:54:21.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/2"}}},"commit":{"parents":[{"commit":"8411c164db8884dbb778d4defdb92369867550c0","subject":"Merge \"trivial: Fix spelling, formatting of vDPA spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8411c164db8884dbb778d4defdb92369867550c0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-05-04 14:53:59.000000000","tz":120},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ba04008b3499dd8877167b30352f90e210b11e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ba04008b3499dd8877167b30352f90e210b11e78"}]},"branch":"refs/heads/master"},"96d197457bbfa8391e8058a0d3d484140e44e104":{"kind":"REWORK","_number":3,"created":"2021-05-12 10:05:33.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/3"}}},"commit":{"parents":[{"commit":"8411c164db8884dbb778d4defdb92369867550c0","subject":"Merge \"trivial: Fix spelling, formatting of vDPA spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8411c164db8884dbb778d4defdb92369867550c0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-05-12 10:05:22.000000000","tz":0},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/96d197457bbfa8391e8058a0d3d484140e44e104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/96d197457bbfa8391e8058a0d3d484140e44e104"}]},"branch":"refs/heads/master"},"fa8464b62970953ac22f88098099e98dba6fa3dd":{"kind":"REWORK","_number":4,"created":"2021-05-13 15:00:11.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/4"}}},"commit":{"parents":[{"commit":"8411c164db8884dbb778d4defdb92369867550c0","subject":"Merge \"trivial: Fix spelling, formatting of vDPA spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8411c164db8884dbb778d4defdb92369867550c0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-05-13 15:00:01.000000000","tz":0},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fa8464b62970953ac22f88098099e98dba6fa3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fa8464b62970953ac22f88098099e98dba6fa3dd"}]},"branch":"refs/heads/master"},"d69bd7e1909d79d204ed18bb5ff2af25ad664f88":{"kind":"REWORK","_number":5,"created":"2021-05-26 13:38:13.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/5"}}},"commit":{"parents":[{"commit":"8411c164db8884dbb778d4defdb92369867550c0","subject":"Merge \"trivial: Fix spelling, formatting of vDPA spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8411c164db8884dbb778d4defdb92369867550c0"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-05-26 13:37:07.000000000","tz":120},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d69bd7e1909d79d204ed18bb5ff2af25ad664f88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d69bd7e1909d79d204ed18bb5ff2af25ad664f88"}]},"branch":"refs/heads/master"},"a22c00d0dc72927dfadb944e3b7c02d3eba5ee21":{"kind":"REWORK","_number":6,"created":"2021-05-28 11:43:10.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/6"}}},"commit":{"parents":[{"commit":"114a7b2a73be6453dc4f97dfb78e64d367f38458","subject":"Merge \"[pps] Move the alternatives to separate section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/114a7b2a73be6453dc4f97dfb78e64d367f38458"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-05-28 11:42:32.000000000","tz":120},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a22c00d0dc72927dfadb944e3b7c02d3eba5ee21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a22c00d0dc72927dfadb944e3b7c02d3eba5ee21"}]},"branch":"refs/heads/master"},"15cce573ff839419f4cc8f1dd23a771070b109e0":{"kind":"REWORK","_number":7,"created":"2021-05-28 11:45:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/43/788243/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/43/788243/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/43/788243/7"}}},"commit":{"parents":[{"commit":"114a7b2a73be6453dc4f97dfb78e64d367f38458","subject":"Merge \"[pps] Move the alternatives to separate section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/114a7b2a73be6453dc4f97dfb78e64d367f38458"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-04-27 10:52:53.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-05-28 11:45:33.000000000","tz":120},"subject":"Allow provider re-parenting in placement","message":"Allow provider re-parenting in placement\n\nThis spec proposes to allow re-parenting and un-parenting\n(or orphaning) RPs via ``PUT /resource_providers/{uuid}``\nAPI in Placement.\n\nStory: 2008764\nTask: 42131\nChange-Id: I5510ebc584842aac5c999fbcce0ff0359016a1f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/15cce573ff839419f4cc8f1dd23a771070b109e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/15cce573ff839419f4cc8f1dd23a771070b109e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
