)]}'
{"id":"openstack%2Fnova-specs~795493","triplet_id":"openstack%2Fnova-specs~master~Id58d2f88c0021c30c8ced90e82fcddf14d310c3d","project":"openstack/nova-specs","branch":"master","topic":"bp/cyborg-admin-user-client","hashtags":[],"change_id":"Id58d2f88c0021c30c8ced90e82fcddf14d310c3d","subject":"Fix the bp link in the cyborg admin token spec","status":"MERGED","created":"2021-06-09 08:00:35.000000000","updated":"2021-06-18 14:01:27.000000000","submitted":"2021-06-18 13:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795493-bp/cyborg-admin-user-client","meta_rev_id":"154ec2370d19e700b6e5bb370f81f45678ec453b","_number":795493,"virtual_id_number":795493,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 13:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-06-18 13:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-06-18 13:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 08:06:11.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-06-09 08:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d36e74e48d2dc130302a9754dedad914641218ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-06-09 08:00:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f65ce6ba872e70a42e5c22c9b088184b50de9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f354b35c3247e5a6058c05be3afa23 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2414cfa543fd4f92a7ec32164656f0bf : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"af4d1122fe437f9a6e363482ce9d1bcb636a094a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-18 13:29:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfast approving as this is a trival url fix","accounts_in_message":[],"_revision_number":1},{"id":"44c14bf4a4e8abbf0412c313996bdd870129a137","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-18 13:29:41.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"20d0a4ce4e8c8fa8405856f28d57f5afc08da782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:29:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeadd327546ff7635ad2dca806531eb140abc2bd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-18 13:32:19.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nas noted on irc when the blueprint is renamed the file should be renamed also\ncyborg-admin-user-client.rst in this case. that is because the move implemented specs script we have uses the file name to find the bluepritn to determin if its completed or not","accounts_in_message":[],"_revision_number":1},{"id":"997565adceecb9b9eddf4c527d873cf1763333c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-06-18 13:33:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c88c741bd8be83fec7f140017e03c064465efd71","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-18 13:35:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\ngreat thanks gibi :)","accounts_in_message":[],"_revision_number":2},{"id":"05905f90834b9fcbf7f1b9e158ad0002c5dc4225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:47:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a37e4de3ac4dd0a6910cb3ab1e68c3 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f823b156e6f141f7aaeb310391b1d987 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b3e5b3afb01859e9fa78ac5e37177ff91c3801e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:47:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79810b541dda57518df00d36809ef2a85b480b24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:59:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/546ff960345a4e2f9228ef2bf0260fc0 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8cdefbdc90416b91d3c3568eb447db : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f58cd174be6c52693cf686735b2aa0367a9b9248","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:59:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"154ec2370d19e700b6e5bb370f81f45678ec453b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 14:01:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/0a42e5f0382545e193fe5bc88726a10f : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45af48ea8dd939a7dce3588b2cf2c0baffae6296","revisions":{"f33529bfe22062d7d8e3a2f70a72fec5c4cf90e6":{"kind":"REWORK","_number":1,"created":"2021-06-09 08:00:35.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/93/795493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/93/795493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/1"}}},"commit":{"parents":[{"commit":"451b2c7e7196045be8b1f27410621f1034450ba9","subject":"Re-propose Unified Limits in Nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/451b2c7e7196045be8b1f27410621f1034450ba9"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-06-09 07:59:47.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-06-09 07:59:47.000000000","tz":120},"subject":"Fix the bp link in the cyborg admin token spec","message":"Fix the bp link in the cyborg admin token spec\n\nBlueprint: cyborg-admin-user-client\nChange-Id: Id58d2f88c0021c30c8ced90e82fcddf14d310c3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f33529bfe22062d7d8e3a2f70a72fec5c4cf90e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f33529bfe22062d7d8e3a2f70a72fec5c4cf90e6"}]},"branch":"refs/heads/master"},"45af48ea8dd939a7dce3588b2cf2c0baffae6296":{"kind":"REWORK","_number":2,"created":"2021-06-18 13:33:20.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/93/795493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/93/795493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/93/795493/2"}}},"commit":{"parents":[{"commit":"451b2c7e7196045be8b1f27410621f1034450ba9","subject":"Re-propose Unified Limits in Nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/451b2c7e7196045be8b1f27410621f1034450ba9"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-06-09 07:59:47.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-06-18 13:32:46.000000000","tz":120},"subject":"Fix the bp link in the cyborg admin token spec","message":"Fix the bp link in the cyborg admin token spec\n\nAlso renamed the spec file to match with the name of the bp.\n\nBlueprint: cyborg-admin-user-client\nChange-Id: Id58d2f88c0021c30c8ced90e82fcddf14d310c3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45af48ea8dd939a7dce3588b2cf2c0baffae6296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45af48ea8dd939a7dce3588b2cf2c0baffae6296"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
