)]}'
{"id":"openstack%2Fnova-specs~799624","triplet_id":"openstack%2Fnova-specs~master~I2b4028afa975b4b11acef62452a8d4dc472ebb2e","project":"openstack/nova-specs","branch":"master","hashtags":[],"change_id":"I2b4028afa975b4b11acef62452a8d4dc472ebb2e","subject":"Add nova-manage commands to show and refresh connection_info","status":"MERGED","created":"2021-07-06 10:12:16.000000000","updated":"2021-07-12 11:15:52.000000000","submitted":"2021-07-12 11:14:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799624","meta_rev_id":"88b4d30a55922e70f6537e3e4a7b2446d1487280","_number":799624,"virtual_id_number":799624,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 11:14:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-07-12 10:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-07-12 11:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-10 14:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-07-12 11:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 10:25:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 12:12:56.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-07-06 16:07:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2021-07-06 16:07:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-09 13:17:20.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"220cec4ab9072b255f6cee994c1632c376f4cd16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-06 10:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03b97f70fb57aaf0b6ea94ae1b2ef6949607663d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:25:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc6440409aa4501a9ef85b6c16cb392 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20aafc1ddc2245028a0f9fc263277452 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"26a5c61a4435c4b65b56fefc01ac1bfa80481ee1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-06 12:12:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)\n\nI have couple of comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"a562a4aaa2ef23028fa07f60aaef51b85dac7c0d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-06 16:07:01.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"48dbf326665a96b20e4efb81cbc4a9735db80dd9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-06 16:07:24.000000000","message":"Patch Set 1: Code-Review-1\n\nGood comments by gibi. I need to load up on more context on this to really grasp things I think","accounts_in_message":[],"_revision_number":1},{"id":"9406915cc4ec95eaab14f63c47010542fc2e1418","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-06 17:25:15.000000000","message":"Patch Set 1:\n\n(11 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c9b10e5a22d34e0c3251d5f5f06cce54a679d6d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-07 16:15:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdb7ed66f7a16816bd2ba662ea12b917dc0a42d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 16:26:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b534b6a90c4b6691f23a43b5fa5d2f : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb516162b5f42d3a73717a05f24c3fb : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"db219754115d1a1a701b10022aa871b34e4cc952","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-09 11:12:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nMy comments were answered. I think it would make sense to lock the instance during the update hence my -1.","accounts_in_message":[],"_revision_number":2},{"id":"f8bda798265ace6f2ffd529fbb92e2bcce4272fc","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-09 11:40:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e29f7494d166aaa2a7dba0ac2290047775b13bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-09 11:44:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b73d78277979afb708cae36f732de4e4d2ef42f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:35:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3b99fb92f4494fa06c2f147e8a4d12 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f86f48ed4a94c40bdbb3eaccaf75b12 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"b01481354d9381058c0fd262743a95e9e6a0e5ab","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-07-09 13:17:20.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)\n\nthis is taking a different approch then i was assumign for the refresh.\ni assume the simple approch i was thinking of will not work?\n\n+1 for now but i dont really have any stong objections to this as proposed.","accounts_in_message":[],"_revision_number":3},{"id":"bba686e7d5d57dc15234c65ab182524ec0060498","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-09 13:29:07.000000000","message":"Patch Set 3:\n\n(4 comments)\n\n\u003e Patch Set 3: Code-Review+1\n\u003e \n\u003e (4 comments)\n\u003e \n\u003e this is taking a different approch then i was assumign for the refresh.\n\u003e i assume the simple approch i was thinking of will not work?\n\u003e \n\u003e +1 for now but i dont really have any stong objections to this as proposed.\n\nYeah GETs aren\u0027t enough to force anything to be regenerated. Most backends will keep connection_info up to date but it\u0027s stuff outside of this like the underlying mapping that would be good to also refresh and fix.","accounts_in_message":[],"_revision_number":3},{"id":"d844486e69196108820ff09b800d40fbee427b09","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-10 14:44:25.000000000","message":"Patch Set 3: Code-Review+2\n\nlooks good","accounts_in_message":[],"_revision_number":3},{"id":"800348d13afd6e5c4fff0d77a9f916a7b29ea987","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-07-12 10:13:16.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nyou answered my question inline so im pretty happy with this.\ngibi seams to be good with this too.\n\nill hold +w for now and see if stephen want to take another look otherwise ill add\n+w after i have checked with him","accounts_in_message":[],"_revision_number":3},{"id":"487c9aa13f8d30a38a270658cba288f9c7bda8b0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 11:00:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fc71e7adea4a3170669967950cd6faa2735031b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:01:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01e7a19a39ba99080b5ba37a798fd1fa08f5dd50","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 11:01:09.000000000","message":"Patch Set 3:\n\nMy comments have been addressed","accounts_in_message":[],"_revision_number":3},{"id":"564f912d1de32777b1e3e1d2e2225f865c9a070c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:14:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6404ae10c8944496b7515f68909f540f : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbb952d182a4ea58bdb65533130645c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fc2b301c894733a00c2a3295651189c7166eb9fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:14:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"88b4d30a55922e70f6537e3e4a7b2446d1487280","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:15:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/7e27f8ba05ca42be9d29026927e8a398 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19f411a07dc99f71f788e6c518482ea1f8e8cebd","revisions":{"57573cb0c5bb366a20deaff7ee423258debe159d":{"kind":"REWORK","_number":1,"created":"2021-07-06 10:12:16.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/24/799624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/799624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/1"}}},"commit":{"parents":[{"commit":"a8c4dffa72801cc0734cdeda8ed65aa992a8fd12","subject":"Merge \"Re-propose \"CPU selection with hypervisor consideration\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a8c4dffa72801cc0734cdeda8ed65aa992a8fd12"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-06-10 07:42:43.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-06 10:12:10.000000000","tz":60},"subject":"Add nova-manage commands to show and refresh connection_info","message":"Add nova-manage commands to show and refresh connection_info\n\nThis spec aims to outline some basic and potentially backportable\n``nova-manage`` commands that will allow operators to both view and\nrefresh the connection_info of a specific block device mapping within\nNova.\n\nBlueprint: nova-manage-refresh-connection-info\nChange-Id: I2b4028afa975b4b11acef62452a8d4dc472ebb2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/57573cb0c5bb366a20deaff7ee423258debe159d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/57573cb0c5bb366a20deaff7ee423258debe159d"}]},"branch":"refs/heads/master"},"cdd3d1e98bbb8ac370b1f4bc1abce7cf0d98c228":{"kind":"REWORK","_number":2,"created":"2021-07-07 16:15:04.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/24/799624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/799624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/2"}}},"commit":{"parents":[{"commit":"54ddb3a6af5b2caf3f38889295e95d6a34620885","subject":"Merge \"Repropose \"vm-boot-with-unaddressed-port\" spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/54ddb3a6af5b2caf3f38889295e95d6a34620885"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-06-10 07:42:43.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-07 16:14:59.000000000","tz":60},"subject":"Add nova-manage commands to show and refresh connection_info","message":"Add nova-manage commands to show and refresh connection_info\n\nThis spec aims to outline some basic and potentially backportable\n``nova-manage`` commands that will allow operators to both view and\nrefresh the connection_info of a specific block device mapping within\nNova.\n\nBlueprint: nova-manage-refresh-connection-info\nChange-Id: I2b4028afa975b4b11acef62452a8d4dc472ebb2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cdd3d1e98bbb8ac370b1f4bc1abce7cf0d98c228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cdd3d1e98bbb8ac370b1f4bc1abce7cf0d98c228"}]},"branch":"refs/heads/master"},"19f411a07dc99f71f788e6c518482ea1f8e8cebd":{"kind":"REWORK","_number":3,"created":"2021-07-09 11:44:41.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/24/799624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/799624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/799624/3"}}},"commit":{"parents":[{"commit":"54ddb3a6af5b2caf3f38889295e95d6a34620885","subject":"Merge \"Repropose \"vm-boot-with-unaddressed-port\" spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/54ddb3a6af5b2caf3f38889295e95d6a34620885"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-06-10 07:42:43.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-09 11:41:00.000000000","tz":60},"subject":"Add nova-manage commands to show and refresh connection_info","message":"Add nova-manage commands to show and refresh connection_info\n\nThis spec aims to outline some basic and potentially backportable\n``nova-manage`` commands that will allow operators to both view and\nrefresh the connection_info of a specific block device mapping within\nNova.\n\nBlueprint: nova-manage-refresh-connection-info\nChange-Id: I2b4028afa975b4b11acef62452a8d4dc472ebb2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/19f411a07dc99f71f788e6c518482ea1f8e8cebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/19f411a07dc99f71f788e6c518482ea1f8e8cebd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
