)]}'
{"id":"openstack%2Fnova-specs~799811","triplet_id":"openstack%2Fnova-specs~master~I54af69453f396a3577503015cd7da7c958d4507a","project":"openstack/nova-specs","branch":"master","topic":"specs/xena/approved/nova-manage-commands-to-move-between-device-models","hashtags":[],"change_id":"I54af69453f396a3577503015cd7da7c958d4507a","subject":"Store and allow libvirt instance device buses and models to be updated","status":"MERGED","created":"2021-07-07 12:48:31.000000000","updated":"2021-07-12 14:18:51.000000000","submitted":"2021-07-12 14:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"799811-specs/xena/approved/nova-manage-commands-to-move-between-device-models","meta_rev_id":"dcce24e8d57509d8d27a1e356b4b6780092daa3f","_number":799811,"virtual_id_number":799811,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 14:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-07-12 13:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-07-12 14:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-12 12:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-07-12 14:04:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 13:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 13:07:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-09 11:24:40.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-07-09 11:57:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"0bce97871ed1cdc97dc757f0a1b478ce7dd776db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-07 12:48:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b5ffbe2277dbd9644f985810e3da2925c9d5d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 13:03:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1399d76a0414ff3aa238170c876bb4a : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc2016ac8df425083b37af9f95dca3b : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5e502934dfd8dc8186a0590eea55848390114a5c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-07 13:07:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nOne nit about naming (yay, bikeshedding) plus a question around when/if we force the hard reboot on the operator.","accounts_in_message":[],"_revision_number":1},{"id":"d4415c760e4a41863e44f47bd75bf4c9fc9a8355","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-07 13:08:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b08faafbb84f1511f65312b349aed4743f5f276","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-07 14:41:52.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"974c4786d5f86b1afb602bc4b2ec14cfcebc9031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-07 17:06:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9148e90a6f7fa164d00b86c4bd47e566e6ee911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 17:17:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e188b06a18914623a63011bdc0d7b015 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a827f609c11a446cb7901db95c4ce109 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d05d5f0860c3a93f5e24ba37a228fc3abf42b8f0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-09 11:24:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)\n\nThis looks pretty strait forward to me.","accounts_in_message":[],"_revision_number":2},{"id":"e173c39d01a16ece92774bb34cf3ba6bc230b184","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-07-09 11:57:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(9 comments)\n\nis because i think you shoudl include a few other models\n\nthhe more genrealised changes could be deferred again but we had this same converation last cycle when we added the machine type changes.\n\ni think just making this general from the start and preventing filed that we know will break would be better long term.","accounts_in_message":[],"_revision_number":2},{"id":"ab12080ca00c3591144ecfc98090e28e1955d9e0","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-09 12:44:07.000000000","message":"Patch Set 2: Workflow-1\n\n(10 comments)\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e (9 comments)\n\u003e \n\u003e is because i think you shoudl include a few other models\n\nACK thanks I got hw_input_bus and hw_pointer_model mixed up.\n\n\u003e thhe more genrealised changes could be deferred again but we had this same converation last cycle when we added the machine type changes.\n\nYeah appreciate that, I still think this more targeted approach is valid here given the need to ensure operators aren\u0027t borking things with specific bus or model updates. For example, trying to use an IDE bus when the underlying instance is using the q35 machine type.\n\n\u003e i think just making this general from the start and preventing filed that we know will break would be better long term.\n\nIMHO it would be much more work to verify. I also hate the try it and see if it works UX and lack specific property validation etc that it would result in. Yeah we can validate the update against o.vo fields but that\u0027s not going to be enough when taking machine types or arches into account etc.\n\nFeel free to write the backlog spec for the generic approach but I\u0027m sticking with this for the time being.","accounts_in_message":[],"_revision_number":2},{"id":"984dcbf6ea1a1da00a0b0b587c8a92e5564a256d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-07-09 13:29:55.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2: Workflow-1\n\u003e \n\u003e (10 comments)\n\u003e \n\u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \n\u003e \u003e (9 comments)\n\u003e \u003e \n\u003e \u003e is because i think you shoudl include a few other models\n\u003e \n\u003e ACK thanks I got hw_input_bus and hw_pointer_model mixed up.\n\u003e \n\u003e \u003e thhe more genrealised changes could be deferred again but we had this same converation last cycle when we added the machine type changes.\n\u003e \n\u003e Yeah appreciate that, I still think this more targeted approach is valid here given the need to ensure operators aren\u0027t borking things with specific bus or model updates. For example, trying to use an IDE bus when the underlying instance is using the q35 machine type.\n\u003e \n\u003e \u003e i think just making this general from the start and preventing filed that we know will break would be better long term.\n\u003e \n\u003e IMHO it would be much more work to verify. I also hate the try it and see if it works UX and lack specific property validation etc that it would result in. Yeah we can validate the update against o.vo fields but that\u0027s not going to be enough when taking machine types or arches into account etc.\n\u003e \n\u003e Feel free to write the backlog spec for the generic approach but I\u0027m sticking with this for the time being.\n\nwe can do the property validation with the generic approch.\nthere is nothing stoping us form doing that its trivial the image porpteis object will do all the validation for us. \n\nso the validation is just read the curent values form the db into the object\ntry to set the filed on the object and if it does nto raise an expction then it was a valid value.\n\n\nwithout calling into the driver code on the destionation comptue and talking to libvirt on that host you will not be able to implenet any better valiation in the nova manage command","accounts_in_message":[],"_revision_number":2},{"id":"c43e42dc425b9be5569fd3c13d2beec83d9d2df9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 11:13:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nTotal bikeshedding, but I\u0027m in agreement with Sean that our command should be named more generically to allow further expansion in the future. All the validation should still take place: we should just remove the specificness from the command names.","accounts_in_message":[],"_revision_number":2},{"id":"bbf50896ace62f303a4cfdd3a95f41d2e394eede","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-07-12 11:43:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbef3ffd4efef2c8bcd4e9d45a2796b43afa5a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-12 12:27:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae37a9fd1599170518515e1fc9aa2b618f099e63","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-12 12:41:53.000000000","message":"Patch Set 3: Code-Review+2\n\nstill looks good. I\u0027m OK with the new, generic command naming.","accounts_in_message":[],"_revision_number":3},{"id":"2a6f6314e7ef9ccb0b54428c0c6bf81cf74d9b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 12:42:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/712b3166780841dba6960c03bdb45c91 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b90aacc1a484660be47bfe015ccf330 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"adeb078e4fed12e715bd78ea2ac471df3438b318","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-07-12 13:12:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)\n\nthis works for me\nlets see what Stephen thinks but this is extensible in the future without needing to modify the command names.","accounts_in_message":[],"_revision_number":3},{"id":"ca163bd45587e139feab4456ffe34e1129f88efe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 14:04:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8eb9797679f11576a4d6dcd04cf17923c49508d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:04:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"73b1382521adaac0cc4aa84fbb7418be486b097d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:17:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8803a507cc4091996cab505b98b524 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf26b375f094d6298652cb3c4ba91d4 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"08146e0ee866137c2e1d45d782f1eb20a000300d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:17:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dcce24e8d57509d8d27a1e356b4b6780092daa3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:18:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/136916810f334aeba0e987093d04221b : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7fbd7eb3179ac0e7b67640ab1e9ad619e6c70b2","revisions":{"85ca28962007b54e4027bc34681eed28a2064531":{"kind":"REWORK","_number":1,"created":"2021-07-07 12:48:31.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/11/799811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/799811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/1"}}},"commit":{"parents":[{"commit":"a8c4dffa72801cc0734cdeda8ed65aa992a8fd12","subject":"Merge \"Re-propose \"CPU selection with hypervisor consideration\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a8c4dffa72801cc0734cdeda8ed65aa992a8fd12"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-07 12:45:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-07 12:47:49.000000000","tz":60},"subject":"Store and allow libvirt instance device buses and models to be updated","message":"Store and allow libvirt instance device buses and models to be updated\n\nChange-Id: I54af69453f396a3577503015cd7da7c958d4507a\nBlueprint: libvirt-device-bus-model-update\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/85ca28962007b54e4027bc34681eed28a2064531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/85ca28962007b54e4027bc34681eed28a2064531"}]},"branch":"refs/heads/master"},"9ff1e894b0a92860257011c9ddc058161a349d1f":{"kind":"REWORK","_number":2,"created":"2021-07-07 17:06:53.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/11/799811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/799811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/2"}}},"commit":{"parents":[{"commit":"a8c4dffa72801cc0734cdeda8ed65aa992a8fd12","subject":"Merge \"Re-propose \"CPU selection with hypervisor consideration\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a8c4dffa72801cc0734cdeda8ed65aa992a8fd12"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-07 12:45:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-07 17:06:24.000000000","tz":60},"subject":"Store and allow libvirt instance device buses and models to be updated","message":"Store and allow libvirt instance device buses and models to be updated\n\nThis spec aims to outline a basic set of nova-manage commands to allow\noperators to move instances between specific device bus and model types\nwithout requiring a rebuild.\n\nChange-Id: I54af69453f396a3577503015cd7da7c958d4507a\nBlueprint: libvirt-device-bus-model-update\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9ff1e894b0a92860257011c9ddc058161a349d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9ff1e894b0a92860257011c9ddc058161a349d1f"}]},"branch":"refs/heads/master"},"e7fbd7eb3179ac0e7b67640ab1e9ad619e6c70b2":{"kind":"REWORK","_number":3,"created":"2021-07-12 12:27:08.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/11/799811/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/11/799811/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/11/799811/3"}}},"commit":{"parents":[{"commit":"4ec721112fc6e04c92fc256e89312c2542ba00b5","subject":"Merge \"Add nova-manage commands to show and refresh connection_info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ec721112fc6e04c92fc256e89312c2542ba00b5"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-07 12:45:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-07-12 12:26:20.000000000","tz":60},"subject":"Store and allow libvirt instance device buses and models to be updated","message":"Store and allow libvirt instance device buses and models to be updated\n\nThis spec aims to outline a basic set of nova-manage commands to allow\noperators to move instances between specific device bus and model types\nwithout requiring a rebuild.\n\nChange-Id: I54af69453f396a3577503015cd7da7c958d4507a\nBlueprint: libvirt-device-bus-model-update\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7fbd7eb3179ac0e7b67640ab1e9ad619e6c70b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e7fbd7eb3179ac0e7b67640ab1e9ad619e6c70b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
