)]}'
{"id":"openstack%2Fnova-specs~803895","triplet_id":"openstack%2Fnova-specs~master~I4caff30f2b3cc12f0970874dcbee04e572e8ccc5","project":"openstack/nova-specs","branch":"master","topic":"configurable-instance-hostnames","hashtags":[],"change_id":"I4caff30f2b3cc12f0970874dcbee04e572e8ccc5","subject":"Revert \"Amend configurable-instance-hostnames to include response changes\"","status":"MERGED","created":"2021-08-09 10:01:57.000000000","updated":"2021-08-19 16:03:23.000000000","submitted":"2021-08-19 16:01:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803895-configurable-instance-hostnames","meta_rev_id":"db722f4953161806363e3685f06e66f7cbd7719a","_number":803895,"virtual_id_number":803895,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 16:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2021-08-19 13:18:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2021-08-09 19:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-08-19 15:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-08-19 15:42:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-09 10:06:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-09 10:06:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-08-09 10:06:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-08-09 10:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 13:18:51.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"}],"messages":[{"id":"55f149fde76bd77d2caa4bb8ca85bf63665c288d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-09 10:01:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a161a9d6267bcd6bc25b04ea2f50595d949af000","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-09 10:06:56.000000000","message":"Patch Set 1:\n\nThis is one of two competing options for this microversion, the other being [1]. From IRC:\n\n\u003cstephenfin\u003e gmann: I\u0027ve proposed two competing spec amendments: one proposes removing all extension prefixes for various \u0027/servers\u0027 and \u0027/flavors\u0027 APIs, and the other proposes not removing any (so \u0027OS-EXT-SRV-ATTR:hostname\u0027 stays the same)\n\u003cstephenfin\u003e gmann: The perfectionist in me would prefer the former. The realist says the latter might be wiser, at least for this microversion.\n\u003cstephenfin\u003e gmann: https://review.opendev.org/c/openstack/nova-specs/+/803892 (remove all prefixes) https://review.opendev.org/c/openstack/nova-specs/+/803895 (remove no prefixes)\n\n[1] https://review.opendev.org/c/openstack/nova-specs/+/803892/","accounts_in_message":[],"_revision_number":1},{"id":"9b4bd37b65e5f62b765399ae9b2d85f5d31fe935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 10:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db16d0c8d774fe499f2cc4115fbc7df : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/920289ca3f5c462db3f89dae0c8a8b2f : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b87148d7f1efd27dde7c7faa02d32dc6bbe273da","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-08-09 19:57:14.000000000","message":"Patch Set 1: Code-Review+2\n\nI am ok with this proposal and discuss the extension prefix removal (proposed in https://review.opendev.org/c/openstack/nova-specs/+/803892) as separate change.","accounts_in_message":[],"_revision_number":1},{"id":"46f430aa99fb4c1e74f32051413c38730f7ca918","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-08-19 13:18:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d9b9ed8121ae5f6bf2e366250b2140208e206b2e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-19 15:42:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd10814316456741bdf53f9f33045f706b35a731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:45:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4f270c07fc561146ca63f1ab52362f24f30a714","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 16:01:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e86964ed4f458fada77b69ae0c1531 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0726885a412b4218bb6ec7090f9df734 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2b8e653d4fe1c446d41b86f1af626460db5bdd52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 16:01:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db722f4953161806363e3685f06e66f7cbd7719a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 16:03:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ebd9a049af6c437b983e1577d7e5f9ea : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed015789ef4c12b95e18366a98837e110ced5776","revisions":{"ed015789ef4c12b95e18366a98837e110ced5776":{"kind":"REWORK","_number":1,"created":"2021-08-09 10:01:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/803895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/803895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/803895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/803895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/803895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/803895/1"}}},"commit":{"parents":[{"commit":"b3877298000356f9ad46ead891f8946278f9a8b9","subject":"Merge \"Move placement specs to placement repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b3877298000356f9ad46ead891f8946278f9a8b9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-09 09:38:53.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-09 10:00:35.000000000","tz":60},"subject":"Revert \"Amend configurable-instance-hostnames to include response changes\"","message":"Revert \"Amend configurable-instance-hostnames to include response changes\"\n\nThis reverts commit aec9a01d54ce02d24745e80cc9d9635a0eeb7048. This\nchange proposed renaming ``OS-EXT-SRV-ATTR:hostname`` attribute in\n``/servers`` responses to match the ``hostname`` attribute in requests,\nwhich would avoid a mismatch between the server requests and server\nresponses. However, this mismatch already exists for the ``host`` and\n``node`` attributes of the server, which are represented as ``host`` and\n``hypervisor_hostname`` in requests (from microversion 2.74) but\n``OS-EXT-SRV-ATTR:host`` and ``OS-EXT-SRV-ATTR:hypervisor_hostname`` in\nresponses. In addition, there are many more fields with the\n``OS-EXT-SRV-ATTR`` prefixes and it seems odd to remove it for one field\n(and force clients to make changes) without removing the other\nextension-based prefixes for both the various ``/servers`` APIs and the\n``/flavors`` APIs. It would be better to tackle this in a separate\nmicroversion or potentially not at all [1].\n\n[1] https://etherpad.opendev.org/p/nova-api-cleanup (Item #3, line 46)\n\nChange-Id: I4caff30f2b3cc12f0970874dcbee04e572e8ccc5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed015789ef4c12b95e18366a98837e110ced5776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed015789ef4c12b95e18366a98837e110ced5776"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
