)]}'
{"id":"openstack%2Fnova-specs~813180","triplet_id":"openstack%2Fnova-specs~master~Iecad5941f3e1e1af6fb774e9cca087c4232169ad","project":"openstack/nova-specs","branch":"master","topic":"specs/xena/approved/manila.rst","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-01-04 14:08:44.000000000","reason":"sean mooney replied on the change"},"10135":{"account":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"last_update":"2022-01-11 12:25:18.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-01-11 12:13:33.000000000","reason":"gibi replied on the change"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2022-01-11 12:25:18.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-01-07 15:15:29.000000000","reason":"Sylvain Bauza replied on the change"}},"hashtags":[],"change_id":"Iecad5941f3e1e1af6fb774e9cca087c4232169ad","subject":"libvirt: Allow Manila shares to be directly attached to instances","status":"MERGED","created":"2021-10-08 12:20:58.000000000","updated":"2022-01-11 14:17:46.000000000","submitted":"2022-01-11 12:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":149,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"813180-specs/xena/approved/manila.rst","meta_rev_id":"1fa70128723aecd64e13f23a7cc8807777f1a05a","_number":813180,"virtual_id_number":813180,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 12:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-01-11 14:17:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-01-11 12:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-07 15:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-01-11 12:13:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-11 14:17:46.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 12:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 15:04:00.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"CC"},{"updated":"2021-10-12 09:27:17.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-10-19 15:53:34.000000000","updated_by":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"reviewer":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"state":"CC"},{"updated":"2021-10-22 13:13:09.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"CC"},{"updated":"2021-10-22 13:56:45.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2021-10-26 13:07:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-12-14 11:44:52.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-12-16 11:43:33.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"6d8ba00f34ba623f96915eb56b25ee5e1cddab42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-10-08 12:20:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3f6b06f170f2662fec05d8423575ec8ddf288b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 12:32:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e32f4250bce4b3fa8824fd845b29872 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fcc296d75a84f7189c58aad490430b4 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ef2c9485477842b7323c6246a0f462794171be24","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-10-12 09:27:17.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"19e430630025a65477a4c703cba3eb02c27212b4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-10-22 13:13:09.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"62fd2dcf8a6bdfc7c85c9146a8392ad13534b3b5","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-10-22 13:24:51.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"09d5224e0fe7bdfbbea7d3397784205da415137f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-10-22 13:28:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9fd027e57eec21b1c7ed2beb38e874ded32d6d9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-10-22 13:36:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb71d18558dd769b4a4b36635aef99e473630146","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-10-22 13:56:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3ce199a7b1ba6634cda269b5a83af1d23778233","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-10-26 13:07:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5029f54ccf5e46546ba0b19fa2539071ff1e282e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-01 11:39:52.000000000","message":"Patch Set 1:\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dcff993f5caa3b64fecfcb76e2db536c036ab33b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-03 11:58:33.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d018ae2dd3fd06d98e7360b802d85ce9b1b03974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-03 12:00:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25607512b486197dfeed86c50cfcf60dc5644480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:16:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91338ef3f494afb9cd508b8f173f22f : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd2710ffe7a74807a476484d0ec61c10 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f188914a76958c5bc42bff77d993ffcfb43a0773","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-11-03 19:23:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d3c24808005cb24d1759c6066015d81067870771","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-04 07:30:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13f348fd112b0f015ec7080d2c2a4c66c281a949","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-04 16:49:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"856e68346d1d190b16c0593b6a1208e417d05751","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-11-16 09:36:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"91f84222a85a274737c7d7dea54a1a0b60d603b4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-11-16 09:37:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50af098345d8db0bdb73d0004e153e9cda08a79e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-17 10:39:56.000000000","message":"Patch Set 2:\n\n(12 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fdc1a5ac0afd8cc1cd2a8d3815fa285488359ed6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-11-18 10:47:05.000000000","message":"Patch Set 2:\n\n(20 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9e483817aa1cb704d9c0d549bb3827693f7707a6","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-24 06:07:57.000000000","message":"Patch Set 2:\n\n(16 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e1aff2d9bb2b16d069650ea57abb77523ce34822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-24 06:18:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3259cc89368e1be8943e0ca0a10372a0322c338c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 06:29:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ba5c8720954b168c13e776d6e4e1a4 : FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c450b53fc5c748d3b58157f78fa78700 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"765b96f21cbaf178f3c187f1afb10ce42b3f87c2","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-24 11:16:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d8e907d5908ac8380b9428098b136bb74cfc46e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-24 11:20:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"296cf6d0d55bf690b257a8eaee8b8ebe806c0862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 11:38:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60f6ee3440a8491ab679e3b35f8eb6c6 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a2305cfbfc4118963d3fc3601fabf9 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":4},{"id":"15db9a07dcee3773647ec8d38fe879e000d29071","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-12-14 10:50:02.000000000","message":"Patch Set 4: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"df709e6b58ec62679a1a922167d08dfe0b8f8b28","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-12-14 11:44:52.000000000","message":"Patch Set 4: Code-Review+1\n\n(11 comments)","accounts_in_message":[],"_revision_number":4},{"id":"40d2465fad4beb82d77245215c2d1c6a6d135246","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-14 20:26:21.000000000","message":"Patch Set 4: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8352cbfd03ab3e0ea019443e3dcf191b53d3574e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 11:43:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ea7956cac532c2a9d079d37852744c0fbea67e4b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2022-01-04 10:53:28.000000000","message":"Patch Set 4:\n\n(13 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bc48313887f74e257bb31437e88778b5bcd24646","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-04 11:24:50.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f8f09d2484367bc2298a57ad1f3fca84d1a1a5f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2022-01-04 13:49:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"15a1c3860718ee721ba20de69dcb276321129abb","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2022-01-04 13:49:49.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5da2ffbbd0fc75529c465cccc0baa0a56c579e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 14:01:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029ef83cb11f473e97acdcc3f4c91e7a : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c3d636c7b34ae3bf8d9c2ccb0f8f03 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":5},{"id":"7602aef5d8344b9fd1bad3cce53e2f94b8ad0a79","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-04 14:08:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"827cccf7021ede2792679ccee3937c769738d37b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-07 15:15:29.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2b888f7b7a96290b1bcf9ef7baebbcb44208d559","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-01-11 12:13:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9297a5a6f127c3ea135aa04559f6aa2d6642a414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:14:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1097dac0119aecd3fe6cf105049b1025318e40e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:25:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd657d45c8904494b3a8ed4edec4a924 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95752ccb4fd2433d90a2935d94e5d3b7 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":5},{"id":"55b09aa9fb9138b067afe887679d00b4e165bf06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:25:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"775c45f2b369a96fc1b252dba5476ed5beed1862","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:28:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/f22df0b6ca6b453e8dbd7011deaf7d1e : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":5},{"id":"9c5878a6c3f0d433d81def74b7aadf92e5219d24","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-11 14:17:46.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82","revisions":{"81dc09f43401f05d4f6436be6f6e1eece8aaec21":{"kind":"REWORK","_number":1,"created":"2021-10-08 12:20:58.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/80/813180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/813180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/1"}}},"commit":{"parents":[{"commit":"e0f91d141c3b8f84261eddec4f3d8462c3aec047","subject":"Update attachment_id spec to include implementation changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0f91d141c3b8f84261eddec4f3d8462c3aec047"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-08 11:21:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-08 12:16:48.000000000","tz":60},"subject":"WIP libvirt: Allow Manila shares to be directly attached to instances","message":"WIP libvirt: Allow Manila shares to be directly attached to instances\n\nChange-Id: Iecad5941f3e1e1af6fb774e9cca087c4232169ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/81dc09f43401f05d4f6436be6f6e1eece8aaec21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/81dc09f43401f05d4f6436be6f6e1eece8aaec21"}]},"branch":"refs/heads/master"},"ae840415359df17060e02e5179b7efe5bf66feeb":{"kind":"REWORK","_number":2,"created":"2021-11-03 12:00:31.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/80/813180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/813180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/2"}}},"commit":{"parents":[{"commit":"c5621dcaae9c26a68fc4edd77eed0947ca5e3430","subject":"Re-propose QoS minimum guaranteed packet rate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c5621dcaae9c26a68fc4edd77eed0947ca5e3430"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-08 11:21:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-11-03 12:00:25.000000000","tz":0},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will outline API,\ndatabase, compute and libvirt driver changes required in Nova to allow the\nshares provided by Manila to be associated with and attached to instances.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: Iecad5941f3e1e1af6fb774e9cca087c4232169ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae840415359df17060e02e5179b7efe5bf66feeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae840415359df17060e02e5179b7efe5bf66feeb"}]},"branch":"refs/heads/master"},"1c775e9746fb5aca003a174111bb23c2bb84ff8e":{"kind":"REWORK","_number":3,"created":"2021-11-24 06:18:48.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/80/813180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/813180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/3"}}},"commit":{"parents":[{"commit":"75579c9917720957cb1b0c565a901ceafc73d946","subject":"fup: Correct the libvirt ephemeral encryption blueprint link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/75579c9917720957cb1b0c565a901ceafc73d946"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-08 11:21:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-11-24 06:13:39.000000000","tz":0},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: Iecad5941f3e1e1af6fb774e9cca087c4232169ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1c775e9746fb5aca003a174111bb23c2bb84ff8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1c775e9746fb5aca003a174111bb23c2bb84ff8e"}]},"branch":"refs/heads/master"},"5d73b6fc21858c9cf113e2922f4ed37739ea0c95":{"kind":"REWORK","_number":4,"created":"2021-11-24 11:20:31.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/80/813180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/813180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/4"}}},"commit":{"parents":[{"commit":"75579c9917720957cb1b0c565a901ceafc73d946","subject":"fup: Correct the libvirt ephemeral encryption blueprint link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/75579c9917720957cb1b0c565a901ceafc73d946"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-08 11:21:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-11-24 11:20:24.000000000","tz":0},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: Iecad5941f3e1e1af6fb774e9cca087c4232169ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5d73b6fc21858c9cf113e2922f4ed37739ea0c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5d73b6fc21858c9cf113e2922f4ed37739ea0c95"}]},"branch":"refs/heads/master"},"2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82":{"kind":"REWORK","_number":5,"created":"2022-01-04 13:49:49.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/80/813180/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/813180/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/813180/5"}}},"commit":{"parents":[{"commit":"29126c44299e07732ff3b8dcea931753b5ae91fc","subject":"Merge \"Add spec for volume backed server rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/29126c44299e07732ff3b8dcea931753b5ae91fc"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-08 11:21:19.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2022-01-04 13:49:33.000000000","tz":0},"subject":"libvirt: Allow Manila shares to be directly attached to instances","message":"libvirt: Allow Manila shares to be directly attached to instances\n\nManila is the OpenStack Shared Filesystems service. This spec will\noutline API, database, compute and libvirt driver changes required in\nNova to allow the shares provided by Manila to be associated with and\nattached to instances.\n\nblueprint: libvirt-virtiofs-attach-manila-shares\nChange-Id: Iecad5941f3e1e1af6fb774e9cca087c4232169ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2cc7d78d7f0f3dcf6bd93afd1e10b76f3848bc82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
