)]}'
{"id":"openstack%2Fnova-specs~815789","triplet_id":"openstack%2Fnova-specs~master~I4903c04ed60bf559a55759b4e0d49728837cb328","project":"openstack/nova-specs","branch":"master","topic":"ironic-instance-change-host","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-18 09:37:07.000000000","reason":"Change was abandoned"},"32281":{"account":{"_account_id":32281,"name":"Simon Li","email":"liboqss2016@163.com","username":"libo"},"last_update":"2022-07-18 09:37:07.000000000","reason":"Change was abandoned"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-07-18 09:36:58.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"hashtags":[],"change_id":"I4903c04ed60bf559a55759b4e0d49728837cb328","subject":"upport ironic instance change host while it\u0027s host not up","status":"ABANDONED","created":"2021-10-28 08:38:22.000000000","updated":"2022-07-18 09:37:07.000000000","total_comment_count":35,"unresolved_comment_count":18,"has_review_started":true,"meta_rev_id":"8f3e3fcb1f1cdbad96e99879383681c164f3c43b","_number":815789,"virtual_id_number":815789,"owner":{"_account_id":32281,"name":"Simon Li","email":"liboqss2016@163.com","username":"libo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-28 08:53:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2021-11-16 17:33:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-07-18 09:36:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 08:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-16 10:08:57.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2021-11-16 10:34:28.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-11-16 17:33:56.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-05-17 01:14:37.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"}],"messages":[{"id":"e6e94cde3146eeb05a70cc2e03f629d371a45bae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32281,"name":"Simon Li","email":"liboqss2016@163.com","username":"libo"},"date":"2021-10-28 08:38:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3dbea594dd10ce069aa2941fba5bfec6ce9352a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:53:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3207e9170fb415b8cbbfbd6d7442569 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00db8a1f73624de3a6d2b05278894565 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e359a938041d3d586e92dc96e360458efaac6f4e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-16 10:08:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"547ed3cd13aa0d66ffcbfd8480f5d83ea783e373","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-11-16 10:34:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c0cfed7cd6a9399625db0b6f8d25bcf97aafaa3b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-11-16 17:33:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87faed8756027b0d2fd6ab56fd916cc7bdf4456a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-12-14 14:50:05.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"80425e27ee1d95997ff370ae743a070c3b7a5589","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-14 09:10:41.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3eee37f12ffcb2623005d24757b771f5513fd1d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-17 01:14:37.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c8dcf728644dcecfb12cb76fec5753f29fd13fd0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-18 09:36:58.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f3e3fcb1f1cdbad96e99879383681c164f3c43b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-18 09:37:07.000000000","message":"Abandoned\n\nAbandoning this open spec as we\u0027re now after next cycle\u0027s spec approval deadline, which means that the proposer hadn\u0027t reproposed this change for this current Zed cycle. Please reopen it again if you want to work on it on the next Aa cycle.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8196adf5bf329a10f039531e202396418ebaaff","revisions":{"c8196adf5bf329a10f039531e202396418ebaaff":{"kind":"REWORK","_number":1,"created":"2021-10-28 08:38:22.000000000","uploader":{"_account_id":32281,"name":"Simon Li","email":"liboqss2016@163.com","username":"libo"},"ref":"refs/changes/89/815789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/815789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/815789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/815789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/815789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/815789/1"}}},"commit":{"parents":[{"commit":"c5621dcaae9c26a68fc4edd77eed0947ca5e3430","subject":"Re-propose QoS minimum guaranteed packet rate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c5621dcaae9c26a68fc4edd77eed0947ca5e3430"}]}],"author":{"name":"libo04","email":"libo04@inspur.com","date":"2021-10-28 08:36:25.000000000","tz":480},"committer":{"name":"libo04","email":"libo04@inspur.com","date":"2021-10-28 08:38:16.000000000","tz":480},"subject":"upport ironic instance change host while it\u0027s host not up","message":"upport ironic instance change host while it\u0027s host not up\n\nCompute service for IronicDriver had been supported multiple host from\nNewton release, but an ironic instance can\u0027t be managed by nova if\nits hypervisor is brought down. Even more serious, an IronicDriver\ncompute usually has a number of ironic hypervisor, while a compute\nservice goes down, many hypervisor go down. Fortunately, The ironic\nhypervisor does not actually create and run instances, so we just need\nto\nre-select a host for ironic instance host that has an up compute\nservice, and change it then we can manage the instance again.\n\nImplements: [blueprint\nironic-instance-change-host](https://blueprints.launchpad.net/nova/+spec/ironic-instance-change-host)\n\nChange-Id: I4903c04ed60bf559a55759b4e0d49728837cb328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c8196adf5bf329a10f039531e202396418ebaaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c8196adf5bf329a10f039531e202396418ebaaff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
