)]}'
{"id":"openstack%2Fnova-specs~821387","triplet_id":"openstack%2Fnova-specs~master~Ic42807e1cb49f86df23106aebd84e78ee5c0f3f5","project":"openstack/nova-specs","branch":"master","topic":"bp/soft-delete-instance-actions","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2021-12-10 14:20:56.000000000","reason":"sean mooney replied on the change"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2021-12-13 11:03:20.000000000","reason":"gibi replied on the change"},"29037":{"account":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"last_update":"2021-12-13 14:16:50.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2021-12-13 14:06:59.000000000","reason":"Sylvain Bauza replied on the change"}},"hashtags":[],"change_id":"Ic42807e1cb49f86df23106aebd84e78ee5c0f3f5","subject":"Implements: blueprint soft-delete-instance-actions","status":"MERGED","created":"2021-12-10 10:43:34.000000000","updated":"2021-12-13 14:20:02.000000000","submitted":"2021-12-13 14:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":15,"has_review_started":true,"submission_id":"821387-bp/soft-delete-instance-actions","meta_rev_id":"08aa4b06d7525c0aa073b89cb8ff08485e35876b","_number":821387,"virtual_id_number":821387,"owner":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-13 14:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2021-12-13 12:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-13 12:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2021-12-13 14:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-12-13 14:04:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-12-13 14:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 10:45:09.000000000","updated_by":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-12-10 10:45:09.000000000","updated_by":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-12-10 10:45:09.000000000","updated_by":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-12-10 10:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"981df95a2f45a65256ec34de947aa1801c1af24a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2021-12-10 10:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b46b0a8bf1722400bd7e0ba387ebd724dacd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 10:49:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97432d5b2fa4ca3904afb62e52e2217 : FAILURE in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bf9a0e1712421ca1c957ce69c5eba5 : FAILURE in 3m 49s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/approved/soft-delete-instance-action.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/approved/soft-delete-instance-action.rst","accounts_in_message":[],"_revision_number":1},{"id":"5733e5afc577556f77545429d15d186633c89ef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2021-12-10 13:35:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9338bcd04917e5cc466e857e9784ed9dff55e4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 13:46:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c673ea276743108439d4e1ebade3c3 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e221775e7f45e3b8d07b939dc0632c : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6cf8d343ce5e418c0035a6f12061752696364af5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-10 14:20:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"29d508bd7e3494e61e5db0de163d3a74a4bc048d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-10 14:21:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7128127bb46e7bd88564ac97c977dcd392880f79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2021-12-13 08:47:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9d73d558b45b551ff72b19d8c99e47dedd65da44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 09:00:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c259e92bf474c83892f046ded30c5e9 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8539a75129684085a8875768af60405b : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"52608d84662f6cd8e344a96e94bb40ff1fd955d1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-12-13 11:03:20.000000000","message":"Patch Set 3: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d7bb07610471ee80e3579cf14f897440133fafd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2021-12-13 12:31:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3e75fe9d916eef83721e0b2accfc3a0a17669d0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-12-13 12:40:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"538dc00961f4f1d91234b1dd335b2e85823c9408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 12:44:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10beabacd2f14add8d255c409150970c : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6faeea8da51443bbc4635636104eb5c : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":4},{"id":"b0ef206f23bb507c0f31ee6743fd89657c58b840","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-13 12:54:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"048682a10058ccff25e9f2f5f8f4472043ac5b51","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-13 14:04:00.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"38ee0c64f62e799070ee58f4628cf0ab30ef151d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 14:04:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3bd5b1d5a579b3ad6533781107249853f3572992","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-12-13 14:06:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c5153a685d3f5ad11bea94fe55c19cff4070700a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 14:16:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c3aee72f7b4eb5bfe7b6e67edb717a : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51a7db89d514761b42c5eaa4668d887 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":4},{"id":"03985953e9aac9c8dacf86d9800cab2c79181632","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 14:16:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"08aa4b06d7525c0aa073b89cb8ff08485e35876b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 14:20:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a74ddf7b1eb34ca3ac1d78bef64e42c2 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3706e912908f102c9afe315fdb37cd1ea24b7eaa","revisions":{"91e5c3fec693d161d753506c46e5ece25af7f62b":{"kind":"REWORK","_number":1,"created":"2021-12-10 10:43:34.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/87/821387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/821387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/1"}}},"commit":{"parents":[{"commit":"fa3654ff192c2e459c0e141e3b45c9b98ea8761e","subject":"Allow project admin to list hypervisors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fa3654ff192c2e459c0e141e3b45c9b98ea8761e"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-10 10:42:04.000000000","tz":60},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-10 10:42:04.000000000","tz":60},"subject":"Implements: blueprint soft-delete-instance-actions","message":"Implements: blueprint soft-delete-instance-actions\n\nChange-Id: Ic42807e1cb49f86df23106aebd84e78ee5c0f3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/91e5c3fec693d161d753506c46e5ece25af7f62b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/91e5c3fec693d161d753506c46e5ece25af7f62b"}]},"branch":"refs/heads/master"},"cf6e99d363fd458da990135328c83e10bd5bbfdf":{"kind":"REWORK","_number":2,"created":"2021-12-10 13:35:29.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/87/821387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/821387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/2"}}},"commit":{"parents":[{"commit":"fa3654ff192c2e459c0e141e3b45c9b98ea8761e","subject":"Allow project admin to list hypervisors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fa3654ff192c2e459c0e141e3b45c9b98ea8761e"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-10 10:42:04.000000000","tz":60},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-10 13:34:31.000000000","tz":60},"subject":"Implements: blueprint soft-delete-instance-actions","message":"Implements: blueprint soft-delete-instance-actions\n\nChange-Id: Ic42807e1cb49f86df23106aebd84e78ee5c0f3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cf6e99d363fd458da990135328c83e10bd5bbfdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cf6e99d363fd458da990135328c83e10bd5bbfdf"}]},"branch":"refs/heads/master"},"3107b8f140cc70586174dec8f3d08085e72b07d8":{"kind":"REWORK","_number":3,"created":"2021-12-13 08:47:48.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/87/821387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/821387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/3"}}},"commit":{"parents":[{"commit":"fa3654ff192c2e459c0e141e3b45c9b98ea8761e","subject":"Allow project admin to list hypervisors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fa3654ff192c2e459c0e141e3b45c9b98ea8761e"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-10 10:42:04.000000000","tz":60},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-13 08:47:34.000000000","tz":60},"subject":"Implements: blueprint soft-delete-instance-actions","message":"Implements: blueprint soft-delete-instance-actions\n\nChange-Id: Ic42807e1cb49f86df23106aebd84e78ee5c0f3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3107b8f140cc70586174dec8f3d08085e72b07d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3107b8f140cc70586174dec8f3d08085e72b07d8"}]},"branch":"refs/heads/master"},"3706e912908f102c9afe315fdb37cd1ea24b7eaa":{"kind":"REWORK","_number":4,"created":"2021-12-13 12:31:59.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/87/821387/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/821387/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/821387/4"}}},"commit":{"parents":[{"commit":"fa3654ff192c2e459c0e141e3b45c9b98ea8761e","subject":"Allow project admin to list hypervisors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fa3654ff192c2e459c0e141e3b45c9b98ea8761e"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-10 10:42:04.000000000","tz":60},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2021-12-13 12:31:35.000000000","tz":60},"subject":"Implements: blueprint soft-delete-instance-actions","message":"Implements: blueprint soft-delete-instance-actions\n\nChange-Id: Ic42807e1cb49f86df23106aebd84e78ee5c0f3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3706e912908f102c9afe315fdb37cd1ea24b7eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3706e912908f102c9afe315fdb37cd1ea24b7eaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
