)]}'
{"id":"openstack%2Fnova-specs~821419","triplet_id":"openstack%2Fnova-specs~master~I84f4db9aa86aec03ce07f27f3e7a39ba2275d351","project":"openstack/nova-specs","branch":"master","topic":"external-cpu-management","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-18 09:36:20.000000000","reason":"Change was abandoned"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2021-12-16 13:33:36.000000000","reason":"Removed by gibi using the hovercard menu"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-01-14 09:10:56.000000000","reason":"Sylvain Bauza replied on the change"}},"hashtags":[],"change_id":"I84f4db9aa86aec03ce07f27f3e7a39ba2275d351","subject":"Add spec for external cpu power managment","status":"ABANDONED","created":"2021-12-10 15:51:50.000000000","updated":"2022-10-17 09:20:18.000000000","total_comment_count":37,"unresolved_comment_count":16,"has_review_started":true,"meta_rev_id":"ebaede4fdf9553d9e55e6875e7533d34896a13f8","_number":821419,"virtual_id_number":821419,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-14 00:12:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2021-12-14 10:00:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-07-18 09:36:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2021-12-14 16:04:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 16:03:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-11 10:22:43.000000000","updated_by":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"reviewer":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"state":"CC"},{"updated":"2021-12-14 00:05:13.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-12-14 10:00:40.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"038f3c7aecf7d602b47ef11c6149dd2c38d01501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-10 15:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab3c877897b83f3257237fc4abed59d5bfe87c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 16:03:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3cd9c94e35b49bbb9ef896d640905ba : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a169ac63034ddab2c03fd6c5defe19 : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ba3816215ce5d6b851ce2e4d1ecc750165c134d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-14 00:03:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"597ba8e7ca94d6cf9f9d6ea2109d71b7a576904a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 00:12:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1ad141b1d5473390fac72262696531 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752b39ebf0a84c5fadbc539d8f81f0db : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"52cf3edf7732ca9080e51b7fc15a5103d031cf08","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-12-14 10:00:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43a5b990aab1c9138ee41dd1f284e22ac5db0b9f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-12-14 15:09:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":2},{"id":"124a7f1ec8b00b9a31c7e798f0a45acb1ac8c9a9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-14 15:59:13.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2a557a17056d9cb69abb1a18f31b4060e95ffbf7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-12-14 16:04:36.000000000","message":"Patch Set 2: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b7b64eb9a05e0934bc95d5a4f98e4724b260c304","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-01-14 09:10:56.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c51b1fe5f69b359506b52ed51b21c7d8e2215e07","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-18 09:36:10.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19d59764da53161227542d390ca9e1e4ddf329bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-18 09:36:20.000000000","message":"Abandoned\n\nAbandoning this open spec as we\u0027re now after next cycle\u0027s spec approval deadline, which means that the proposer hadn\u0027t reproposed this change for this current Zed cycle. Please reopen it again if you want to work on it on the next Aa cycle.","accounts_in_message":[],"_revision_number":2},{"id":"ebaede4fdf9553d9e55e6875e7533d34896a13f8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-10-17 09:20:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b241b8673b22441890beccd3fa3c75016e10d655","revisions":{"95542b7025bd00b1e47a96067024db934a7f5dba":{"kind":"REWORK","_number":1,"created":"2021-12-10 15:51:50.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/19/821419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/19/821419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/1"}}},"commit":{"parents":[{"commit":"516e90fa6355412ff05cffd8f322d15dda1a507a","subject":"Merge \"Re-propose Remove tenant_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/516e90fa6355412ff05cffd8f322d15dda1a507a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-12-10 15:51:16.000000000","tz":0},"subject":"Add spec for external cpu power managment","message":"Add spec for external cpu power managment\n\nChange-Id: I84f4db9aa86aec03ce07f27f3e7a39ba2275d351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95542b7025bd00b1e47a96067024db934a7f5dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95542b7025bd00b1e47a96067024db934a7f5dba"}]},"branch":"refs/heads/master"},"b241b8673b22441890beccd3fa3c75016e10d655":{"kind":"REWORK","_number":2,"created":"2021-12-14 00:03:41.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/19/821419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/19/821419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/19/821419/2"}}},"commit":{"parents":[{"commit":"516e90fa6355412ff05cffd8f322d15dda1a507a","subject":"Merge \"Re-propose Remove tenant_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/516e90fa6355412ff05cffd8f322d15dda1a507a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-12-10 15:51:16.000000000","tz":0},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2021-12-14 00:03:41.000000000","tz":0},"subject":"Add spec for external cpu power managment","message":"Add spec for external cpu power managment\n\nChange-Id: I84f4db9aa86aec03ce07f27f3e7a39ba2275d351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b241b8673b22441890beccd3fa3c75016e10d655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b241b8673b22441890beccd3fa3c75016e10d655"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
